)]}'
{"id":"jjb%2Fpython-jenkins~657628","triplet_id":"jjb%2Fpython-jenkins~master~I7ca1f2e15a6c40883ba376b182c2c991ca76e8b2","project":"jjb/python-jenkins","branch":"master","topic":"bug/1828051","hashtags":[],"change_id":"I7ca1f2e15a6c40883ba376b182c2c991ca76e8b2","subject":"Make get_job_info fetch_all_builds work with jobs in folders","status":"MERGED","created":"2019-05-07 15:27:10.000000000","updated":"2019-05-23 14:11:46.000000000","submitted":"2019-05-23 14:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657628-1558620706810-263a428c","meta_rev_id":"000ec307d7ee599671a81d8b9d720c7aeb00a6dc","_number":657628,"virtual_id_number":657628,"owner":{"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"date":"2019-05-08 07:53:57.000000000","_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},{"value":0,"date":"2019-05-07 15:27:24.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"date":"2019-05-08 07:42:37.000000000","_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},{"value":2,"date":"2019-05-23 14:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":2,"date":"2019-05-23 13:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-23 13:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-23 13:47:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},{"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 15:27:24.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-05-08 07:53:57.000000000","updated_by":{"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},"reviewer":{"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},"state":"REVIEWER"},{"updated":"2019-05-23 13:42:55.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-05-23 13:47:24.000000000","updated_by":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"reviewer":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"state":"REVIEWER"},{"updated":"2019-05-23 14:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c876973e7e26a1f81eef701a672df1c556c67c8e","author":{"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},"date":"2019-05-07 15:27:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a34ec21e8fd87d7167c5629dbe93a64386c623c5","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-05-07 15:27:24.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"839d165efd4a5a50ad7df3b7994b1c674bd19cac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 15:44:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/28/657628/1/check/openstack-tox-cover/df4239d/cover/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/28/657628/1/check/openstack-tox-pep8/f635993/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/28/657628/1/check/openstack-tox-py27/4a28f55/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/28/657628/1/check/openstack-tox-py35/1b9af7c/ : SUCCESS in 3m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/28/657628/1/check/build-openstack-sphinx-docs/7eee543/html/ : SUCCESS in 3m 51s\n- your-readthedocs-job-requires-attention http://logs.openstack.org/28/657628/1/check/your-readthedocs-job-requires-attention/7318516/ : FAILURE in 38s (non-voting)\n- openstack-tox-pypy http://logs.openstack.org/28/657628/1/check/openstack-tox-pypy/4d0b086/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/28/657628/1/check/openstack-tox-docs/e24f873/html/ : SUCCESS in 2m 57s\n- openstack-tox-linters http://logs.openstack.org/28/657628/1/check/openstack-tox-linters/d1ad32d/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/28/657628/1/check/openstack-tox-py36/b00fa3e/ : SUCCESS in 3m 19s\n- jjb-tox-cross-jenkins-job-builder http://logs.openstack.org/28/657628/1/check/jjb-tox-cross-jenkins-job-builder/c094b4c/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e82402edc4b2a970f66986bb3f4840bf2715bb10","author":{"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},"date":"2019-05-07 19:07:12.000000000","message":"Patch Set 1:\n\nI don\u0027t currently have access to any such Jenkins instance, but I think that the fullName fields isn\u0027t guaranteed to always be there. If you have a Jenkins without folder support, it\u0027ll likely be missing. Therefore we might need to fallback to name.","accounts_in_message":[],"_revision_number":1},{"id":"04c6f3c7d9fb8f8b33490b56dfe66195ed4f7b59","author":{"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},"date":"2019-05-08 07:42:37.000000000","message":"Patch Set 1:\n\n\u003e I don\u0027t currently have access to any such Jenkins instance, but I\n \u003e think that the fullName fields isn\u0027t guaranteed to always be there.\n \u003e If you have a Jenkins without folder support, it\u0027ll likely be\n \u003e missing. Therefore we might need to fallback to name.\n\nI just installed a fresh Jenkins 2.164.2 with no plugins. fullName was still present in get_job_info. Obviously this might not be true for all older Jenkins versions.","accounts_in_message":[],"_revision_number":1},{"id":"96a6f3ffc02e726a471beaf7916ea2c0466d7bca","author":{"_account_id":27651,"name":"Tomas Janousek","email":"tomi@nomi.cz","username":"tomas.janousek"},"date":"2019-05-08 07:53:57.000000000","message":"Patch Set 1:\n\nHi,\n\n \u003e I just installed a fresh Jenkins 2.164.2 with no plugins. fullName\n \u003e was still present in get_job_info. Obviously this might not be true\n \u003e for all older Jenkins versions.\n\nOh, in that case it might just be fine. I used to work in a company with an old Jenkins, but I\u0027m no longer there and they\u0027ve since upgraded, so this might not be an issue after all. Thanks for checking on a fresh and clean Jenkins! I\u0027ll take a closer look at the code later. :-)","accounts_in_message":[],"_revision_number":1},{"id":"d3daaea141986803ae406bba1a8880d3f13c6504","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 13:42:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef903f3f6d2a817ab7900df16aafa9fc32546e99","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2019-05-23 13:47:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6853d10bca41a243a9dc9060194dece87f6c754","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 13:47:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24a143c5f7071b26e826acca358252515a98500e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:11:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/28/657628/1/gate/build-openstack-sphinx-docs/e3876db/html/ : SUCCESS in 2m 59s\n- openstack-tox-py35 http://logs.openstack.org/28/657628/1/gate/openstack-tox-py35/a17b42b/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/28/657628/1/gate/openstack-tox-py36/801fa7d/ : SUCCESS in 2m 42s\n- openstack-tox-linters http://logs.openstack.org/28/657628/1/gate/openstack-tox-linters/69f5a88/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/28/657628/1/gate/openstack-tox-py27/5f5d461/ : SUCCESS in 3m 26s\n- jjb-tox-cross-jenkins-job-builder http://logs.openstack.org/28/657628/1/gate/jjb-tox-cross-jenkins-job-builder/55d65d5/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d5ed1f6c606e697e8320744c65f31fcfa3b00a0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:11:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"190fbbb5ad04e97c4450fe7c78afe638b0d9e5e3","revisions":{"190fbbb5ad04e97c4450fe7c78afe638b0d9e5e3":{"kind":"REWORK","_number":1,"created":"2019-05-07 15:27:10.000000000","uploader":{"_account_id":30390,"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","username":"flichtenheld"},"ref":"refs/changes/28/657628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/28/657628/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/28/657628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/28/657628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/28/657628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/28/657628/1"}}},"commit":{"parents":[{"commit":"a5615d04f87f5b5dca7adfb07ea6aae29574192c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a5615d04f87f5b5dca7adfb07ea6aae29574192c"}]}],"author":{"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","date":"2019-05-07 14:55:48.000000000","tz":120},"committer":{"name":"Frank Lichtenheld","email":"frank.lichtenheld@sophos.com","date":"2019-05-07 15:23:14.000000000","tz":120},"subject":"Make get_job_info fetch_all_builds work with jobs in folders","message":"Make get_job_info fetch_all_builds work with jobs in folders\n\nMake sure to use fullName from the API results to correctly\ncompute the URL for the additional API request.\n\nChange-Id: I7ca1f2e15a6c40883ba376b182c2c991ca76e8b2\nCloses-Bug:#1828051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/190fbbb5ad04e97c4450fe7c78afe638b0d9e5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/190fbbb5ad04e97c4450fe7c78afe638b0d9e5e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
