)]}'
{"id":"jjb%2Fpython-jenkins~659131","triplet_id":"jjb%2Fpython-jenkins~master~I7dfd92c011b223d171ffd4be9c3f8c59fe87c6cf","project":"jjb/python-jenkins","branch":"master","topic":"mock","hashtags":[],"change_id":"I7dfd92c011b223d171ffd4be9c3f8c59fe87c6cf","subject":"Remove pin on mock module","status":"MERGED","created":"2019-05-14 17:57:32.000000000","updated":"2019-05-23 14:11:46.000000000","submitted":"2019-05-23 14:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659131-1558620706066-a6855eb4","meta_rev_id":"cf39abe9b211d85466effa8843471d99b55272e7","_number":659131,"virtual_id_number":659131,"owner":{"_account_id":6039,"name":"Joni Harker","email":"code@gentlydownthe.net","username":"jesusaurus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2019-05-23 14:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1054,"name":"Darragh Bailey","email":"daragh.bailey@gmail.com","username":"dbailey-k"},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":2,"date":"2019-05-23 13:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1054,"name":"Darragh Bailey","email":"daragh.bailey@gmail.com","username":"dbailey-k"},{"value":2,"date":"2019-05-23 13:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-23 13:04:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1054,"name":"Darragh Bailey","email":"daragh.bailey@gmail.com","username":"dbailey-k"},{"value":1,"date":"2019-05-23 13:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1054,"name":"Darragh Bailey","email":"daragh.bailey@gmail.com","username":"dbailey-k"},{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 13:04:21.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-05-23 13:04:59.000000000","updated_by":{"_account_id":1054,"name":"Darragh Bailey","email":"daragh.bailey@gmail.com","username":"dbailey-k"},"reviewer":{"_account_id":1054,"name":"Darragh Bailey","email":"daragh.bailey@gmail.com","username":"dbailey-k"},"state":"REVIEWER"},{"updated":"2019-05-23 13:46:13.000000000","updated_by":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"reviewer":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"state":"REVIEWER"},{"updated":"2019-05-23 14:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2edf85482af0433f56e26303fdc5c82908329c1","author":{"_account_id":6039,"name":"Joni Harker","email":"code@gentlydownthe.net","username":"jesusaurus"},"date":"2019-05-14 17:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4172ebef40795eec6c8c59c23bbda3305833e6f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:05:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/31/659131/1/check/openstack-tox-cover/d41c233/cover/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/31/659131/1/check/openstack-tox-pep8/8f987a1/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/31/659131/1/check/openstack-tox-py27/21689b4/ : FAILURE in 2m 10s\n- openstack-tox-py35 http://logs.openstack.org/31/659131/1/check/openstack-tox-py35/14ebde0/ : SUCCESS in 3m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/31/659131/1/check/build-openstack-sphinx-docs/a32b5a2/html/ : SUCCESS in 2m 41s\n- your-readthedocs-job-requires-attention http://logs.openstack.org/31/659131/1/check/your-readthedocs-job-requires-attention/bcadcb6/ : FAILURE in 34s (non-voting)\n- openstack-tox-pypy http://logs.openstack.org/31/659131/1/check/openstack-tox-pypy/dd3b5c3/ : FAILURE in 2m 37s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/31/659131/1/check/openstack-tox-docs/d7d4712/html/ : SUCCESS in 2m 46s\n- openstack-tox-linters http://logs.openstack.org/31/659131/1/check/openstack-tox-linters/ea60f4f/ : SUCCESS in 2m 39s\n- openstack-tox-py36 http://logs.openstack.org/31/659131/1/check/openstack-tox-py36/b543bf1/ : SUCCESS in 3m 02s\n- jjb-tox-cross-jenkins-job-builder http://logs.openstack.org/31/659131/1/check/jjb-tox-cross-jenkins-job-builder/5ea59ad/ : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6b68ecb67cd6e37ab52fb3983d04d0ee421bd7c0","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 11:38:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"befa76b4eb1b7ab3987a7ccdd386e55e14f07ebc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 12:03:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/31/659131/2/check/build-openstack-sphinx-docs/00080a0/html/ : SUCCESS in 3m 12s\n- your-readthedocs-job-requires-attention http://logs.openstack.org/31/659131/2/check/your-readthedocs-job-requires-attention/040281c/ : FAILURE in 33s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/31/659131/2/check/openstack-tox-cover/fd961be/cover/ : SUCCESS in 5m 56s\n- openstack-tox-py35 http://logs.openstack.org/31/659131/2/check/openstack-tox-py35/cce4b88/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/31/659131/2/check/openstack-tox-py36/cd8a897/ : SUCCESS in 5m 34s\n- openstack-tox-linters http://logs.openstack.org/31/659131/2/check/openstack-tox-linters/c2ad7f7/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/31/659131/2/check/openstack-tox-py27/5714285/ : SUCCESS in 5m 46s\n- openstack-tox-pypy http://logs.openstack.org/31/659131/2/check/openstack-tox-pypy/d8bcc1c/ : SUCCESS in 4m 10s\n- jjb-tox-cross-jenkins-job-builder http://logs.openstack.org/31/659131/2/check/jjb-tox-cross-jenkins-job-builder/ee6370e/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"06f7316ceb1442bb1e434f3736c75ec61fc2edcb","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 13:04:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"48c6a76e147a9f660ac3bd3d69da39d9e10d2799","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2019-05-23 13:46:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8dc05dd151645b1e81d04e596d3bac3dd20760ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 13:46:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef5dd89dad318c20141b7c21e40e99753cd8f216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:11:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/31/659131/2/gate/build-openstack-sphinx-docs/77857a8/html/ : SUCCESS in 2m 47s\n- openstack-tox-py35 http://logs.openstack.org/31/659131/2/gate/openstack-tox-py35/cf55c07/ : SUCCESS in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/31/659131/2/gate/openstack-tox-py36/363cceb/ : SUCCESS in 3m 37s\n- openstack-tox-linters http://logs.openstack.org/31/659131/2/gate/openstack-tox-linters/58f80ab/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/31/659131/2/gate/openstack-tox-py27/e86a2f3/ : SUCCESS in 3m 13s\n- jjb-tox-cross-jenkins-job-builder http://logs.openstack.org/31/659131/2/gate/jjb-tox-cross-jenkins-job-builder/ce4925d/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"501f3196c47706deb4b2d40ec5bbd831518f5089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:11:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ab91c7a7c2e5e0ed5661497f2d47e7ef4d95ecd","revisions":{"e40929ff8e51f6d53756b174116458d0347a6ddb":{"kind":"REWORK","_number":1,"created":"2019-05-14 17:57:32.000000000","uploader":{"_account_id":6039,"name":"Joni Harker","email":"code@gentlydownthe.net","username":"jesusaurus"},"ref":"refs/changes/31/659131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/31/659131/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/1"}}},"commit":{"parents":[{"commit":"a5615d04f87f5b5dca7adfb07ea6aae29574192c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a5615d04f87f5b5dca7adfb07ea6aae29574192c"}]}],"author":{"name":"K Jonathan Harker","email":"jharker@suse.com","date":"2019-05-14 17:55:31.000000000","tz":-420},"committer":{"name":"K Jonathan Harker","email":"jharker@suse.com","date":"2019-05-14 17:55:31.000000000","tz":-420},"subject":"Remove pin on mock module","message":"Remove pin on mock module\n\nThis was pinned for python26, which we no longer support.\n\nChange-Id: I7dfd92c011b223d171ffd4be9c3f8c59fe87c6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/e40929ff8e51f6d53756b174116458d0347a6ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/e40929ff8e51f6d53756b174116458d0347a6ddb"}]},"branch":"refs/heads/master"},"0ab91c7a7c2e5e0ed5661497f2d47e7ef4d95ecd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 11:38:13.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/31/659131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/31/659131/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/31/659131/2"}}},"commit":{"parents":[{"commit":"748aa2068bf0052a61adff61326a4896bae61b66","subject":"Update jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/748aa2068bf0052a61adff61326a4896bae61b66"}]}],"author":{"name":"K Jonathan Harker","email":"jharker@suse.com","date":"2019-05-14 17:55:31.000000000","tz":-420},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-05-23 11:38:13.000000000","tz":0},"subject":"Remove pin on mock module","message":"Remove pin on mock module\n\nThis was pinned for python26, which we no longer support.\n\nChange-Id: I7dfd92c011b223d171ffd4be9c3f8c59fe87c6cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/0ab91c7a7c2e5e0ed5661497f2d47e7ef4d95ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/0ab91c7a7c2e5e0ed5661497f2d47e7ef4d95ecd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
