)]}'
{"id":"jjb%2Fpython-jenkins~682235","triplet_id":"jjb%2Fpython-jenkins~master~Ifb8d10414130f9322886d242d42713074c06fb6b","project":"jjb/python-jenkins","branch":"master","hashtags":[],"change_id":"Ifb8d10414130f9322886d242d42713074c06fb6b","subject":"two new methods: upsert_job and check_jenkinsfile_syntax","status":"ABANDONED","created":"2019-09-15 04:39:51.000000000","updated":"2019-09-18 12:20:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2d826bfcb12d97caa8d5e9b1eaaecc5c2e0e970","_number":682235,"virtual_id_number":682235,"owner":{"_account_id":30978,"name":"Marcos Diez","email":"marcos@unitron.com.br","username":"mdiez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2019-09-15 04:49:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-15 04:40:02.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-09-15 04:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56d8d9f8e1c96c1fc6924aa3a2ad17558bc4e673","author":{"_account_id":30978,"name":"Marcos Diez","email":"marcos@unitron.com.br","username":"mdiez"},"date":"2019-09-15 04:39:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ddf4a5c8f66f4fd69c6cf740cd13fbe8d04b874","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-09-15 04:40:02.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"afdaca70df42deaea2dcf8988d5a51177df19cdd","author":{"_account_id":30978,"name":"Marcos Diez","email":"marcos@unitron.com.br","username":"mdiez"},"date":"2019-09-15 04:40:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8d72721a3006bf6ed5c7a8f27fd2eaceba3bce2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 04:49:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d4c3c584c8d7485bac1bfeaeb53b3a6e : SUCCESS in 3m 16s\n- your-readthedocs-job-requires-attention https://zuul.opendev.org/t/openstack/build/f34065954fbf4418b23ee8871083933a : FAILURE in 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8adea6125cfc4ce789c9bc3c73c537e5 : SUCCESS in 3m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/244b1481cb8f4899b8909cedd88eb2f1 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc952de940ba44b4b6cb0e0df4097862 : SUCCESS in 3m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e10e7f57528403681aadf5251397beb : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efe6c092dff04f66a8d5746acac4640d : SUCCESS in 2m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/04daef889b0c42d69584c232d78ca987 : SUCCESS in 3m 49s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/46eebceff9fa48938189d746e5c98d34 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d2d826bfcb12d97caa8d5e9b1eaaecc5c2e0e970","author":{"_account_id":30978,"name":"Marcos Diez","email":"marcos@unitron.com.br","username":"mdiez"},"date":"2019-09-18 12:20:04.000000000","message":"Abandoned\n\nAs requested, I replaced by 3 other PRs, with individual changes!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7cfb04cabd2fcaf3c4fafb28821ad12c0c22b92","revisions":{"5195b97f74efd4457623920c6db70c87a520a5ca":{"kind":"REWORK","_number":1,"created":"2019-09-15 04:39:51.000000000","uploader":{"_account_id":30978,"name":"Marcos Diez","email":"marcos@unitron.com.br","username":"mdiez"},"ref":"refs/changes/35/682235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/35/682235/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/1"}}},"commit":{"parents":[{"commit":"35f660a8b95e4ef1df7edcaf429085842b6fbae5","subject":"Merge \"When updating jobs the response body may be empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/35f660a8b95e4ef1df7edcaf429085842b6fbae5"}]}],"author":{"name":"Marcos Diez","email":"marcos@unitron.com.br","date":"2019-09-15 04:17:26.000000000","tz":-180},"committer":{"name":"Marcos Diez","email":"marcos@unitron.com.br","date":"2019-09-15 04:39:05.000000000","tz":-180},"subject":"two new methos: upsert_job and check_jenkinsfile_syntax","message":"two new methos: upsert_job and check_jenkinsfile_syntax\n\nChange-Id: Ifb8d10414130f9322886d242d42713074c06fb6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5195b97f74efd4457623920c6db70c87a520a5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5195b97f74efd4457623920c6db70c87a520a5ca"}]},"branch":"refs/heads/master"},"a7cfb04cabd2fcaf3c4fafb28821ad12c0c22b92":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-15 04:40:41.000000000","uploader":{"_account_id":30978,"name":"Marcos Diez","email":"marcos@unitron.com.br","username":"mdiez"},"ref":"refs/changes/35/682235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/35/682235/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/35/682235/2"}}},"commit":{"parents":[{"commit":"35f660a8b95e4ef1df7edcaf429085842b6fbae5","subject":"Merge \"When updating jobs the response body may be empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/35f660a8b95e4ef1df7edcaf429085842b6fbae5"}]}],"author":{"name":"Marcos Diez","email":"marcos@unitron.com.br","date":"2019-09-15 04:17:26.000000000","tz":-180},"committer":{"name":"Marcos Diez","email":"marcos@unitron.com.br","date":"2019-09-15 04:40:11.000000000","tz":-180},"subject":"two new methods: upsert_job and check_jenkinsfile_syntax","message":"two new methods: upsert_job and check_jenkinsfile_syntax\n\nChange-Id: Ifb8d10414130f9322886d242d42713074c06fb6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a7cfb04cabd2fcaf3c4fafb28821ad12c0c22b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a7cfb04cabd2fcaf3c4fafb28821ad12c0c22b92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
