)]}'
{"id":"jjb%2Fpython-jenkins~694980","triplet_id":"jjb%2Fpython-jenkins~master~I20c23395d89035aae14463718688d227e0b2bef4","project":"jjb/python-jenkins","branch":"master","topic":"enabled_nested_view","hashtags":[],"change_id":"I20c23395d89035aae14463718688d227e0b2bef4","subject":"fix bug with nested views","status":"ABANDONED","created":"2019-11-19 10:30:07.000000000","updated":"2023-07-24 11:34:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b1a1c76de2761053983eae66043b098bd728878","_number":694980,"virtual_id_number":694980,"owner":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-19 12:01:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 10:30:19.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-11-19 12:01:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb8ef5520375617fcf90a14c753c4188594e0774","author":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"date":"2019-11-19 10:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a95ece5cccca8039ad01446a82ea0685840391a4","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-11-19 10:30:19.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"bf2c5b017fff13eb6772aee4c6fb5de7223ee3ea","author":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"date":"2019-11-19 10:35:45.000000000","message":"Topic changed from enabled nested view to enabled_nested_view","accounts_in_message":[],"_revision_number":1},{"id":"389d24f9418f63ef0beb2113b47462692aa88397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:42:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab5a750cd2540a28eb583aa95800fb8 : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2ab122ce7894b96976341f32ded1bf3 : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8e66247e31e94b9d9ddd6d22c68af5b1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb092e2d650d40fbaa0047fae4e783c3 : SUCCESS in 3m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f068a4efda1d41d4ac4653999b662207 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23e5ec1e06a44588aa789f0400ae53a0 : SUCCESS in 4m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e694135428884884a6344497408d48bb : SUCCESS in 6m 32s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/403f9cd0f0a44f0b9ac27c9d604ed604 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5c4ff73d7fc0c205c5bb55bcd30e81f875e54db0","author":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"date":"2019-11-19 11:51:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d301d7c3a1edefb0307e1b9d6b49cf5de43787d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 12:01:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375d1c07eee14491b4f6cc280ff1cbae : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e53bcd2bc6a4dfe9581b0ad44056b52 : FAILURE in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/379233c066ff4502944ca52440edb8b1 : FAILURE in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e417df7651904aaa8024d47429f6bd69 : FAILURE in 3m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8834b1430dc042a38b72adeb779c9ae3 : FAILURE in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8359095f2c7c4c64a6cee2ec94158458 : FAILURE in 3m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7f8cf175926c4c56847c15f07ad5ac90 : FAILURE in 4m 24s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/a7636a82069d4de2a694bb0a51f09b13 : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1b1a1c76de2761053983eae66043b098bd728878","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"date":"2023-07-24 11:34:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dbdf80b99910a1a4e9a7d75e0cb50a3ae2f6bcfd","revisions":{"ed4c0b5e790f1921335c8313f5a2029b8bb2f201":{"kind":"REWORK","_number":1,"created":"2019-11-19 10:30:07.000000000","uploader":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"ref":"refs/changes/80/694980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/80/694980/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/1"}}},"commit":{"parents":[{"commit":"1749d4d2f54c7e7d12a0d893e772030a727923d1","subject":"Update docs building job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/1749d4d2f54c7e7d12a0d893e772030a727923d1"}]}],"author":{"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","date":"2019-11-19 10:30:07.000000000","tz":0},"committer":{"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","date":"2019-11-19 10:30:07.000000000","tz":0},"subject":"Change the default path in _get_job_folder","message":"Change the default path in _get_job_folder\n\nChange-Id: I20c23395d89035aae14463718688d227e0b2bef4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/ed4c0b5e790f1921335c8313f5a2029b8bb2f201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/ed4c0b5e790f1921335c8313f5a2029b8bb2f201"}]},"branch":"refs/heads/master"},"dbdf80b99910a1a4e9a7d75e0cb50a3ae2f6bcfd":{"kind":"REWORK","_number":2,"created":"2019-11-19 11:51:53.000000000","uploader":{"_account_id":31307,"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","username":"fhiekel"},"ref":"refs/changes/80/694980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/80/694980/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/80/694980/2"}}},"commit":{"parents":[{"commit":"1749d4d2f54c7e7d12a0d893e772030a727923d1","subject":"Update docs building job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/1749d4d2f54c7e7d12a0d893e772030a727923d1"}]}],"author":{"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","date":"2019-11-19 11:48:48.000000000","tz":60},"committer":{"name":"Friedrich Hiekel","email":"friedrich.hiekel@cloudandheat.com","date":"2019-11-19 11:51:05.000000000","tz":60},"subject":"fix bug with nested views","message":"fix bug with nested views\n\nChange-Id: I20c23395d89035aae14463718688d227e0b2bef4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/dbdf80b99910a1a4e9a7d75e0cb50a3ae2f6bcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/dbdf80b99910a1a4e9a7d75e0cb50a3ae2f6bcfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
