)]}'
{"id":"jjb%2Fpython-jenkins~700196","triplet_id":"jjb%2Fpython-jenkins~master~I30351ec78d6186b0674e3c08be00076b0ce2333e","project":"jjb/python-jenkins","branch":"master","hashtags":[],"change_id":"I30351ec78d6186b0674e3c08be00076b0ce2333e","subject":"Include placeholder tasks in get_running_builds","status":"MERGED","created":"2019-12-20 16:28:41.000000000","updated":"2020-03-04 02:58:33.000000000","submitted":"2020-03-04 02:58:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700196-1583290713119-30632db2","meta_rev_id":"60a1017e2884998ef669438b610bfcfa9ec5ba99","_number":700196,"virtual_id_number":700196,"owner":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"date":"2020-01-01 18:27:44.000000000","_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 02:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-26 21:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-04 02:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 16:28:54.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-03-04 02:50:04.000000000","updated_by":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"reviewer":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"state":"REVIEWER"},{"updated":"2020-03-04 02:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86dbf83d3c5cf96f8bf3149ad6891d375f94e0c8","author":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"date":"2019-12-20 16:28:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfbe93107e5860134188430dc1c6cfbde0c6cb5f","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-12-20 16:28:54.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"1de593e836efb9b630d9b03539e7c3c1821a62f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 16:40:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b534e68540164263a911af8d79ce509b : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b109d2bec15422b97938989bddb7dc4 : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/968407e0ee31416d8c047f0a2ec46e23 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3f63f917cf14cde9b8d037edcfffe38 : SUCCESS in 4m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/568fd9e43d1b4a19a3c55389ec3a3e47 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3077542b0a3249729c13368faba25075 : SUCCESS in 5m 37s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/55f46c9907654ceeb6ba774a909b0203 : SUCCESS in 4m 33s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/2a4421c51aa94429ade50acd091521ec : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2ed3ab3ac0d22d0b82c0fa16cdca49325c8a7e6c","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2019-12-22 13:56:25.000000000","message":"Patch Set 1: Code-Review-1\n\nThe commit message doesn\u0027t really explain why this is needed or what the benefit of it is. Can you add more details to the commit message?","accounts_in_message":[],"_revision_number":1},{"id":"6b73e1ce4b599b54c7778bcdfa3be01a8cc54c4e","author":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"date":"2019-12-25 07:42:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f50a9c3ad98ff99b02db7d07f34aba058750aa78","author":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"date":"2019-12-25 07:44:35.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"05c9d6303133baeb811f041bb5f31666898661c6","author":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"date":"2019-12-25 07:48:27.000000000","message":"Patch Set 3:\n\nHi Thanh Ha,\nthanks for your input! I\u0027ve added more details to the commit message. I also noticed that I had mixed up the bug ids. I\u0027ve added Related-Bug and fixed Closes-Bug in the commit message.","accounts_in_message":[],"_revision_number":3},{"id":"0d7f5a5ccdc7ee5fbeddadf606c74303419fb175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 07:53:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7887029fb38b4f9f86d5e9c32b88c3d2 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9227797685cf4ec0b252196c2492940f : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ec1b0e65a3e74f73890862f2946531f9 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c3528736284fe5aa9fef850fbf8154 : SUCCESS in 4m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/edca1345fb594a78b2c025927341785b : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e139fe9744ff4d3395ba67e0f42e6d27 : SUCCESS in 5m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7fbf042b76b74f6aadb06f8ca77a8257 : SUCCESS in 5m 00s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/8c7601a607db4ac0ae3be1ba6810099f : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f628620a4c0ed9c01d145ec0c73ada5f431fcc4b","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2019-12-26 21:23:40.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for adding clarity to the commit message. This makes more sense to me now.","accounts_in_message":[],"_revision_number":3},{"id":"24a0a4a3fe4ba550146c53bed7718ef5a2925954","author":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"date":"2020-01-01 18:27:44.000000000","message":"Patch Set 3:\n\nWhat is required to get this merged? Is there anything missing from my side?","accounts_in_message":[],"_revision_number":3},{"id":"1388bbd6858d5a80b31bbfb2778a31bd01b9f206","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2020-01-01 19:32:54.000000000","message":"Patch Set 3:\n\n\u003e What is required to get this merged? Is there anything missing from\n \u003e my side?\n\nNothing on your end. Typically we prefer to have 2 reviewers review a patch before it\u0027s merged. Worst case if no one else reviews the patch by the time I decide to cut a release I will merge any outstanding reviews that only have 1 +2 before we release.","accounts_in_message":[],"_revision_number":3},{"id":"75496e578c7d671a7680b14b8da4549c37251999","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2020-03-04 02:50:04.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"de3c17a632281a3ddd9ac46af25f995e4ea66944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 02:50:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a6b8935564cdd399862a9b3c1866f204d9a12387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 02:58:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0cbf2269b6449139d094fb24df512cc : SUCCESS in 3m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5087cd70cabe43bcba0cceca63b0d678 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d080fd6c314d4b27b6da9148f8f4ff36 : SUCCESS in 3m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b535119d01fc4d54afcab17dfe72f642 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aeadab776bd946bd8a229cdcec86a376 : SUCCESS in 3m 15s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/600636a385114a989e0c84eb75afac90 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"60a1017e2884998ef669438b610bfcfa9ec5ba99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 02:58:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19e7f7395aaa0d876a39baf82a80b775d17ff2b3","revisions":{"ccc97d3df4f6cea1e566ada7e307dc03ef0c301a":{"kind":"REWORK","_number":1,"created":"2019-12-20 16:28:41.000000000","uploader":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"ref":"refs/changes/96/700196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/96/700196/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/1"}}},"commit":{"parents":[{"commit":"36c99d34366aa9b72b86a920fae138865b953d79","subject":"Merge \"new method: create_folder (with tests)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/36c99d34366aa9b72b86a920fae138865b953d79"}]}],"author":{"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","date":"2019-12-20 16:08:26.000000000","tz":60},"committer":{"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","date":"2019-12-20 16:28:13.000000000","tz":60},"subject":"Include placeholder tasks in get_running_builds","message":"Include placeholder tasks in get_running_builds\n\nChange-Id: I30351ec78d6186b0674e3c08be00076b0ce2333e\nCloses-Bug: #1659787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/ccc97d3df4f6cea1e566ada7e307dc03ef0c301a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/ccc97d3df4f6cea1e566ada7e307dc03ef0c301a"}]},"branch":"refs/heads/master"},"354d3a1e15f9a395e25890ab21a18293f4855bca":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-25 07:42:37.000000000","uploader":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"ref":"refs/changes/96/700196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/96/700196/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/2"}}},"commit":{"parents":[{"commit":"36c99d34366aa9b72b86a920fae138865b953d79","subject":"Merge \"new method: create_folder (with tests)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/36c99d34366aa9b72b86a920fae138865b953d79"}]}],"author":{"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","date":"2019-12-20 16:08:26.000000000","tz":60},"committer":{"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","date":"2019-12-25 07:32:59.000000000","tz":60},"subject":"Include placeholder tasks in get_running_builds","message":"Include placeholder tasks in get_running_builds\n\nPipeline jobs in in Jenkins 2.x appear as placeholder tasks for their\nentire lifetime when queried via the API. The fix for #1659787\nintroduced completly ignoring placeholder tasks. Thus pipeline jobs are\nnever included in the list returned by get_running_builds.\n\nThis commit will change the behaviour of get_running_builds as follows:\n  - Placeholder tasks will be included: So running pipeline jobs\n    will be included.\n  - Builds which do not have a build number yet will be ignored:\n    This prevents the exception described in #1659787.\n\nChange-Id: I30351ec78d6186b0674e3c08be00076b0ce2333e\nCloses-Bug: #1659787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/354d3a1e15f9a395e25890ab21a18293f4855bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/354d3a1e15f9a395e25890ab21a18293f4855bca"}]},"branch":"refs/heads/master"},"19e7f7395aaa0d876a39baf82a80b775d17ff2b3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-25 07:44:35.000000000","uploader":{"_account_id":31448,"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","username":"pascalhofmann"},"ref":"refs/changes/96/700196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/96/700196/3","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/96/700196/3"}}},"commit":{"parents":[{"commit":"36c99d34366aa9b72b86a920fae138865b953d79","subject":"Merge \"new method: create_folder (with tests)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/36c99d34366aa9b72b86a920fae138865b953d79"}]}],"author":{"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","date":"2019-12-20 16:08:26.000000000","tz":60},"committer":{"name":"Pascal Hofmann","email":"mail@pascalhofmann.de","date":"2019-12-25 07:43:46.000000000","tz":60},"subject":"Include placeholder tasks in get_running_builds","message":"Include placeholder tasks in get_running_builds\n\nPipeline jobs in in Jenkins 2.x appear as placeholder tasks for their\nentire lifetime when queried via the API. The fix for #1659787\nintroduced completly ignoring placeholder tasks. Thus pipeline jobs are\nnever included in the list returned by get_running_builds.\n\nThis commit will change the behaviour of get_running_builds as follows:\n  - Placeholder tasks will be included: So running pipeline jobs\n    will be included.\n  - Builds which do not have a build number yet will be ignored:\n    This prevents the exception described in #1659787.\n\nChange-Id: I30351ec78d6186b0674e3c08be00076b0ce2333e\nCloses-Bug: #1750372\nRelated-Bug: #1659787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/19e7f7395aaa0d876a39baf82a80b775d17ff2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/19e7f7395aaa0d876a39baf82a80b775d17ff2b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
