)]}'
{"id":"jjb%2Fpython-jenkins~743072","triplet_id":"jjb%2Fpython-jenkins~master~I264010f2125831b1699b9d343a4de118bb7db972","project":"jjb/python-jenkins","branch":"master","topic":"bug/1872942","hashtags":[],"change_id":"I264010f2125831b1699b9d343a4de118bb7db972","subject":"Switch links for contributors to opendev.org or launchpad","status":"MERGED","created":"2020-07-26 20:14:57.000000000","updated":"2020-08-04 07:10:21.000000000","submitted":"2020-08-04 07:10:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743072-1596525021878-3ec634e1","meta_rev_id":"8f291e9b4c8bfb1862417bfe94683e65084cd178","_number":743072,"virtual_id_number":743072,"owner":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 07:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"all":[{"value":0,"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},{"value":2,"date":"2020-08-04 06:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-01 12:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"all":[{"value":0,"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},{"value":1,"date":"2020-08-04 06:57:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-26 21:41:29.000000000","updated_by":{"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},"reviewer":{"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},"state":"REVIEWER"},{"updated":"2020-08-01 12:59:16.000000000","updated_by":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"reviewer":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"state":"REVIEWER"},{"updated":"2020-08-04 06:57:33.000000000","updated_by":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"reviewer":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"state":"REVIEWER"},{"updated":"2020-08-04 07:10:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1046af253525de537e92ccdb2b926f23be68d85b","author":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"date":"2020-07-26 20:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61c1dc04bfd8550957a4a51c3bddad1b0ae14857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 20:24:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa4ce1aaaac54aa8af43307d1fb04dcb : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9079ef2ac70e4edb8385009c92d4c56b : SUCCESS in 3m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8739b14ad4654ad496d191902ccedf16 : SUCCESS in 4m 06s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/c34327900e7e4d96bd8c60d8c087bdcd : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e428a5142158048745d93689badd0c4e05971543","author":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"date":"2020-07-26 20:26:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23cd54a59c43713b1d3bc59949e172194e9bd775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 20:37:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b29cecc82e64f129a36b1aad34ad31c : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb521ecf22fc42a786c6e62ce0eacdb7 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b20010df4dff489dafb7e262da3cb262 : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bc37ec24470245518edaad9537f57e64 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/011e5af39084483fa3b6a2cf9365c9c9 : SUCCESS in 3m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4a43f0a504ab4be2be6d8660c4549fd7 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57dee5f951aa48e2b1f3a82231f75d88 : SUCCESS in 4m 54s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/6ad3c265a101482aa1695dcf7ac911a2 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"dbe0146a93a066dac51ece53ae7e52ebf4d62c8d","author":{"_account_id":215,"name":"Jon-Paul Sullivan","email":"j3p0uk@hotmail.com","username":"jonpaul-sullivan"},"date":"2020-07-26 21:41:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f1303ea5b1c84c83bb607187b38a8412f8c6f8fc","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2020-07-27 10:20:05.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThere is one more to update in README.rst\n\n\nI also spotted the old openstack URL in tox.ini:\n\n tox.ini:28:    pip install -q -U -e \u0027git+https://git.openstack.org/openstack-infra/jenkins-job-builder@master#egg\u003djenkins-job-builder\u0027 ; fi \"\n\nWhich can be changed to:\n\n git+https://opendev.org/jjb/python-jenkins@master#egg\u003djenkins-job-builder\n\n:)","accounts_in_message":[],"_revision_number":2},{"id":"6b15f5fe315a309d78a257b4a471019ad291578e","author":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"date":"2020-08-01 07:28:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a98d10738cef68f594853e7803628843288590f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 07:41:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5cf33768d614c42a8724ffd5ff14b04 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1780f117befe4bd7b2f79166fccb7252 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbe44b8ced3449a9959c74cda61d8da3 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2af4518ffabf4cf4806cd484f7870eff : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38cc8cf221384e53a931a8a2bbe33a1a : SUCCESS in 3m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60bdf329c2714cb8a0415a47a2739515 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7274a6b683c4490b2240828783cf1a1 : SUCCESS in 3m 16s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/11bdba36b1c1464d97c8df743b88f17b : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"d22f9eb17270092daa0de5cd0092a3e45aa5c859","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2020-08-01 12:59:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"849b44d79df19c06aec7731a0436077f7e7a1140","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2020-08-04 06:57:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you Marcin!","accounts_in_message":[],"_revision_number":3},{"id":"5e6995d77b9ddbb87ca2dde49e3ffb3ca808eb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 06:57:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0344d25ac5a231ba498299a62e54b35b4546685c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 07:10:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f7151c63e92ab8f236677114ceeda633cfa60aa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 07:10:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6cfe6be7adc04833b58baf68471faacc : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322dda17fc4148ffa92a865c4398b4ef : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f970ab507bef431c8e99fd12a086af0a : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ebb1282af6e4b0595700dbcc7665c06 : SUCCESS in 4m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce4df6bbb64d4d31a7843b1b180af567 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0274b24945b4bef996f51bf40a84569 : SUCCESS in 5m 21s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/94fdab7a70fe452fad9cb7162e65f31b : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"34254af2b0a1f002e20b59bf70e64faaa6bf540b","revisions":{"de9d7119cbd0ba14f7cff767a63113235dbd9dbe":{"kind":"REWORK","_number":1,"created":"2020-07-26 20:14:57.000000000","uploader":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"ref":"refs/changes/72/743072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/72/743072/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/1"}}},"commit":{"parents":[{"commit":"4aab17d404f019b2eaf4eadd3d7bf993576b9935","subject":"Add retrieval of stages and artifacts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/4aab17d404f019b2eaf4eadd3d7bf993576b9935"}]}],"author":{"name":"Marcin Cieślak","email":"saper@saper.info","date":"2020-07-26 19:49:51.000000000","tz":120},"committer":{"name":"Marcin Cieślak","email":"saper@saper.info","date":"2020-07-26 20:14:31.000000000","tz":120},"subject":"Switch links for contributors to opendev.org","message":"Switch links for contributors to opendev.org\n\nChange-Id: I264010f2125831b1699b9d343a4de118bb7db972\nCloses-Bug: #1872942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/de9d7119cbd0ba14f7cff767a63113235dbd9dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/de9d7119cbd0ba14f7cff767a63113235dbd9dbe"}]},"branch":"refs/heads/master"},"b0f9e24f808ea8ca61c135d64b0d21891d93c0a2":{"kind":"REWORK","_number":2,"created":"2020-07-26 20:26:47.000000000","uploader":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"ref":"refs/changes/72/743072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/72/743072/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/2"}}},"commit":{"parents":[{"commit":"4aab17d404f019b2eaf4eadd3d7bf993576b9935","subject":"Add retrieval of stages and artifacts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/4aab17d404f019b2eaf4eadd3d7bf993576b9935"}]}],"author":{"name":"Marcin Cieślak","email":"saper@saper.info","date":"2020-07-26 19:49:51.000000000","tz":120},"committer":{"name":"Marcin Cieślak","email":"saper@saper.info","date":"2020-07-26 20:25:43.000000000","tz":120},"subject":"Switch links for contributors to opendev.org or launchpad","message":"Switch links for contributors to opendev.org or launchpad\n\nAlso change the mailing list address\n\nChange-Id: I264010f2125831b1699b9d343a4de118bb7db972\nCloses-Bug: #1872942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/b0f9e24f808ea8ca61c135d64b0d21891d93c0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/b0f9e24f808ea8ca61c135d64b0d21891d93c0a2"}]},"branch":"refs/heads/master"},"34254af2b0a1f002e20b59bf70e64faaa6bf540b":{"kind":"REWORK","_number":3,"created":"2020-08-01 07:28:42.000000000","uploader":{"_account_id":2961,"name":"Marcin Cieślak","email":"saper@saper.info","username":"saperski"},"ref":"refs/changes/72/743072/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/72/743072/3","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/72/743072/3"}}},"commit":{"parents":[{"commit":"4aab17d404f019b2eaf4eadd3d7bf993576b9935","subject":"Add retrieval of stages and artifacts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/4aab17d404f019b2eaf4eadd3d7bf993576b9935"}]}],"author":{"name":"Marcin Cieślak","email":"saper@saper.info","date":"2020-07-26 19:49:51.000000000","tz":120},"committer":{"name":"Marcin Cieślak","email":"saper@saper.info","date":"2020-08-01 07:27:53.000000000","tz":120},"subject":"Switch links for contributors to opendev.org or launchpad","message":"Switch links for contributors to opendev.org or launchpad\n\nAlso change the mailing list address\n\nChange-Id: I264010f2125831b1699b9d343a4de118bb7db972\nCloses-Bug: #1872942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/34254af2b0a1f002e20b59bf70e64faaa6bf540b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/34254af2b0a1f002e20b59bf70e64faaa6bf540b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
