)]}'
{"id":"jjb%2Fpython-jenkins~802098","triplet_id":"jjb%2Fpython-jenkins~master~I3c36d65437ff5c26b20976d8a2ac05245947968a","project":"jjb/python-jenkins","branch":"master","topic":"bug/1687551","hashtags":[],"change_id":"I3c36d65437ff5c26b20976d8a2ac05245947968a","subject":"fix-bug1687551-url-too-long","status":"NEW","created":"2021-07-23 19:04:43.000000000","updated":"2021-07-25 14:20:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9abeea21b2bcddf14d7f2e4d356ba745b717a141","_number":802098,"virtual_id_number":802098,"owner":{"_account_id":33792,"name":"liuxun","display_name":"tsymq00","email":"tsymq00@gmail.com","username":"tsymq00"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-25 14:20:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 19:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-24 02:57:59.000000000","updated_by":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"reviewer":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"state":"REVIEWER"}],"messages":[{"id":"3b305a6692f4df4c78a1f271d3a08065aa1376ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33792,"name":"liuxun","display_name":"tsymq00","email":"tsymq00@gmail.com","username":"tsymq00"},"date":"2021-07-23 19:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06d99970ee33157d3b80d848464a2a2e39e405a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 19:18:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee56319dd2a4ea8ade875234e0cc0d3 : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/624a2cbc67c7451a8ce1abbcc37ac9a7 : FAILURE in 6m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/41d1d1085ad8493db1bf664d8f15de89 : FAILURE in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e856bb0f3cdd44979227d78502b05bf7 : FAILURE in 4m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f533daf07fa545d8bcb54c2820f8d3a3 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64fb66b71d9146249b8daf3b417d6632 : FAILURE in 8m 14s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/75fb7902be054783833bb5a880c46414 : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"55dea1fd55f8735c335e57fabbe89264898689f2","author":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"date":"2021-07-24 02:57:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce45a4bc241e6f8b12797a7ec5bfd1d00156987d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33792,"name":"liuxun","display_name":"tsymq00","email":"tsymq00@gmail.com","username":"tsymq00"},"date":"2021-07-25 14:10:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9abeea21b2bcddf14d7f2e4d356ba745b717a141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 14:20:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef3a534f904c4b0b8e67c42c5ad0617b : FAILURE in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/140743e594eb4413b025274bcf4fd46d : FAILURE in 4m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6a49be8c75124211948a7299e62cdd2c : FAILURE in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5af2ecf625334e328f1004b11526b462 : FAILURE in 5m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7af7e91101014306be01a526ab54e8d7 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12a119f6df31415897df9407030cbf3d : FAILURE in 5m 07s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/7bc04cb48dd944d88d8f71ca574d4262 : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"334d681802af826d82fc21ff4e783e2b3efcfc03","revisions":{"5b3b1fcb7a189a6b50556add767ea4480b793c79":{"kind":"REWORK","_number":1,"created":"2021-07-23 19:04:43.000000000","uploader":{"_account_id":33792,"name":"liuxun","display_name":"tsymq00","email":"tsymq00@gmail.com","username":"tsymq00"},"ref":"refs/changes/98/802098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/98/802098/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/1"}}},"commit":{"parents":[{"commit":"570a143c74d092efaf9bc68a3bae9b15804f4a63","subject":"Allow build number to be a string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/570a143c74d092efaf9bc68a3bae9b15804f4a63"}]}],"author":{"name":"tsymq00","email":"tsymq00@gmail.com","date":"2021-07-23 14:53:03.000000000","tz":480},"committer":{"name":"tsymq00","email":"tsymq00@gmail.com","date":"2021-07-23 19:04:11.000000000","tz":480},"subject":"fix-bug1687551-url-too-long","message":"fix-bug1687551-url-too-long\n\nIn case of large parameters, only encode (optional) token in url\nquery string to avoid HTTP 414 ERROR: URI Too Long.\n\nCloses-Bug: 1687551\nChange-Id: I3c36d65437ff5c26b20976d8a2ac05245947968a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5b3b1fcb7a189a6b50556add767ea4480b793c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5b3b1fcb7a189a6b50556add767ea4480b793c79"}]},"branch":"refs/heads/master"},"334d681802af826d82fc21ff4e783e2b3efcfc03":{"kind":"REWORK","_number":2,"created":"2021-07-25 14:10:56.000000000","uploader":{"_account_id":33792,"name":"liuxun","display_name":"tsymq00","email":"tsymq00@gmail.com","username":"tsymq00"},"ref":"refs/changes/98/802098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/98/802098/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/98/802098/2"}}},"commit":{"parents":[{"commit":"570a143c74d092efaf9bc68a3bae9b15804f4a63","subject":"Allow build number to be a string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/570a143c74d092efaf9bc68a3bae9b15804f4a63"}]}],"author":{"name":"tsymq00","email":"tsymq00@gmail.com","date":"2021-07-23 14:53:03.000000000","tz":480},"committer":{"name":"tsymq00","email":"tsymq00@gmail.com","date":"2021-07-25 14:09:18.000000000","tz":480},"subject":"fix-bug1687551-url-too-long","message":"fix-bug1687551-url-too-long\n\nmodify jenkins.Jenkins.build_job_url:\nIn case of large parameters, we can set ``encode_parameters_in_url``(which\nis True by default for forward compatibility) to False to avoid HTTP\n414 ERROR: URI Too Long.\n\nCloses-Bug: 1687551\nChange-Id: I3c36d65437ff5c26b20976d8a2ac05245947968a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/334d681802af826d82fc21ff4e783e2b3efcfc03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/334d681802af826d82fc21ff4e783e2b3efcfc03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
