)]}'
{"id":"jjb%2Fpython-jenkins~821781","triplet_id":"jjb%2Fpython-jenkins~master~I97ba8ca004387b376a511ba52fc9fdd2aeb98a63","project":"jjb/python-jenkins","branch":"master","topic":"bug/1685575","attention_set":{},"removed_from_attention_set":{"34368":{"account":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"last_update":"2021-12-15 08:10:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I97ba8ca004387b376a511ba52fc9fdd2aeb98a63","subject":"Use fullname in get_job_info_regex","status":"MERGED","created":"2021-12-14 23:35:02.000000000","updated":"2021-12-15 08:10:42.000000000","submitted":"2021-12-15 08:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821781-bug/1685575","meta_rev_id":"e9e8d146318bc8170b555bbf3d1eeeb9f2652b99","_number":821781,"virtual_id_number":821781,"owner":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 08:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"all":[{"value":2,"date":"2021-12-15 08:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-15 01:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"all":[{"value":1,"date":"2021-12-15 08:02:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-14 23:48:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 01:02:14.000000000","updated_by":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"reviewer":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"state":"REVIEWER"},{"updated":"2021-12-15 08:02:43.000000000","updated_by":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"reviewer":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"state":"REVIEWER"}],"messages":[{"id":"50543c3fdadf80480b566f6d4f8795ca9a59a4cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"date":"2021-12-14 23:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23d4b6d37e97c526bfd5cc2b421847025b450eb4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"date":"2021-12-14 23:35:25.000000000","message":"Topic fix-1685575-regex-fullname removed","accounts_in_message":[],"_revision_number":1},{"id":"7c1feb33e4e8a964304e807a71e3e1274a11c4b9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"date":"2021-12-14 23:35:46.000000000","message":"Topic set to bug/1685575","accounts_in_message":[],"_revision_number":1},{"id":"b1d5439149cfb181571f03acefef07488ea03b15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"date":"2021-12-14 23:39:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"de5438d2cffb4dd511a106ca6750690695d38858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 23:48:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8539533cf74dfa925f54366ff5a647 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d18b0800480c4865ac2784a36b511093 : SUCCESS in 3m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8d01584790d24132856d4e49c29a32f6 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/286a0d1376334a4495aaa3fbd983ed58 : SUCCESS in 3m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f482ee6b8bbf49e0bde862b176aac702 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0058b1e695f44cafb75d9e15b3862ca5 : SUCCESS in 4m 31s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/daead19406ae482b8a0a92f9697b0dc2 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"16dc88b62747e29e914d33185fc4fdc374d1d3e2","author":{"_account_id":13807,"name":"Thanh Ha","display_name":"zxiiro","email":"zxiiro@gmail.com","username":"zxiiro"},"date":"2021-12-15 01:02:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd48760c120ba260ece91840987f3fd860438f28","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2021-12-15 08:02:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bf1d86ebd25f05e9fd84e57eab44730bdd166e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 08:03:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce9cdedad30e4deb963a9a816ed8ab160618a1a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 08:10:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca885fb6f504a8c9a66ee14a21c4df2 : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/56e81d9d1c6546029b47fb49525c9e37 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98af87058d5c438e86a17ce498ad4c04 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a91c850f9044b0abe2caf11e2de9c05 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/560fb30a4f8240349f345b0a63eff1a5 : SUCCESS in 5m 42s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/68cb159b08fe45158d410b9d20ece696 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e9e8d146318bc8170b555bbf3d1eeeb9f2652b99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 08:10:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dc2b9447de6fefa716dab5c64a197b55ca461fc5","revisions":{"0872ed1e80f631433343a15b7e8b5824a6cc286a":{"kind":"REWORK","_number":1,"created":"2021-12-14 23:35:02.000000000","uploader":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"ref":"refs/changes/81/821781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/81/821781/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/1"}}},"commit":{"parents":[{"commit":"570a143c74d092efaf9bc68a3bae9b15804f4a63","subject":"Allow build number to be a string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/570a143c74d092efaf9bc68a3bae9b15804f4a63"}]}],"author":{"name":"Jim Wisniewski","email":"wisnij@gmail.com","date":"2021-12-14 23:17:53.000000000","tz":-300},"committer":{"name":"Jim Wisniewski","email":"wisnij@gmail.com","date":"2021-12-14 23:33:18.000000000","tz":-300},"subject":"Use fullname in get_job_info_regex","message":"Use fullname in get_job_info_regex\n\nUse full job name in `get_job_info_regex` instead of just the name, so\nthat matches will work in folders as well.\n\nChange-Id: I97ba8ca004387b376a511ba52fc9fdd2aeb98a63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/0872ed1e80f631433343a15b7e8b5824a6cc286a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/0872ed1e80f631433343a15b7e8b5824a6cc286a"}]},"branch":"refs/heads/master"},"dc2b9447de6fefa716dab5c64a197b55ca461fc5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-14 23:39:03.000000000","uploader":{"_account_id":34368,"name":"Jim Wisniewski","email":"wisnij@gmail.com","username":"wisnij"},"ref":"refs/changes/81/821781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/81/821781/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/81/821781/2"}}},"commit":{"parents":[{"commit":"570a143c74d092efaf9bc68a3bae9b15804f4a63","subject":"Allow build number to be a string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/570a143c74d092efaf9bc68a3bae9b15804f4a63"}]}],"author":{"name":"Jim Wisniewski","email":"wisnij@gmail.com","date":"2021-12-14 23:17:53.000000000","tz":-300},"committer":{"name":"Jim Wisniewski","email":"wisnij@gmail.com","date":"2021-12-14 23:39:03.000000000","tz":0},"subject":"Use fullname in get_job_info_regex","message":"Use fullname in get_job_info_regex\n\nUse full job name in `get_job_info_regex` instead of just the name, so\nthat matches will work in folders as well.\n\nCloses-Bug: 1685575\nChange-Id: I97ba8ca004387b376a511ba52fc9fdd2aeb98a63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/dc2b9447de6fefa716dab5c64a197b55ca461fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/dc2b9447de6fefa716dab5c64a197b55ca461fc5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
