)]}'
{"id":"jjb%2Fpython-jenkins~832169","triplet_id":"jjb%2Fpython-jenkins~master~I1d07e5a6b73dffbd132859c8dbe9d03f1d4678a8","project":"jjb/python-jenkins","branch":"master","hashtags":[],"change_id":"I1d07e5a6b73dffbd132859c8dbe9d03f1d4678a8","subject":"CI test","status":"ABANDONED","created":"2022-03-06 13:51:56.000000000","updated":"2022-03-07 12:12:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"16184250f6825d233da9a1bed58312fdbeed1693","_number":832169,"virtual_id_number":832169,"owner":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-06 13:58:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-03-06 13:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1659c7b0bcd64a8e8f03a7acd6af759ec8ca503b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"date":"2022-03-06 13:51:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac25f88a24a8523bcb7c44139cad9cc76503889b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"date":"2022-03-06 13:52:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"db8f48deb41d2cd3f227531563ae0220babcb6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-06 13:58:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb99573712b487ab1c0722c0e1e2ac9 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a92d6b81fbc54808bb8ced567714255c : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/64c1c5a7d1614fd694071457f6eaeec6 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99142a4bd16c4e8db5a857200ed63269 : SUCCESS in 5m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/89479d55587544ea8ac6557826effc7c : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5179d925b81a457facc771ca7d038265 : SUCCESS in 5m 09s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/800f254174b741e198418b49a3d2fc1f : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"16184250f6825d233da9a1bed58312fdbeed1693","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"date":"2022-03-07 12:12:15.000000000","message":"Abandoned\n\nIt was only a test","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc443418dc59a0882f0fce358b247dd5c8e58bc0","revisions":{"fc443418dc59a0882f0fce358b247dd5c8e58bc0":{"kind":"REWORK","_number":1,"created":"2022-03-06 13:51:56.000000000","uploader":{"_account_id":31965,"name":"Adam Romanek","email":"romanek.adam@gmail.com","username":"romanek-adam"},"ref":"refs/changes/69/832169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/69/832169/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/69/832169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/69/832169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/69/832169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/69/832169/1"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Adam Romanek","email":"romanek.adam@gmail.com","date":"2022-03-06 13:49:45.000000000","tz":60},"committer":{"name":"Adam Romanek","email":"romanek.adam@gmail.com","date":"2022-03-06 13:51:46.000000000","tz":60},"subject":"CI test","message":"CI test\n\nChange-Id: I1d07e5a6b73dffbd132859c8dbe9d03f1d4678a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/fc443418dc59a0882f0fce358b247dd5c8e58bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/fc443418dc59a0882f0fce358b247dd5c8e58bc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
