)]}'
{"id":"jjb%2Fpython-jenkins~877252","triplet_id":"jjb%2Fpython-jenkins~master~I52aa4f3af07d6d14044fa54939ffea928051f953","project":"jjb/python-jenkins","branch":"master","hashtags":[],"change_id":"I52aa4f3af07d6d14044fa54939ffea928051f953","subject":"Replace pkg_resources with packaging module","status":"ABANDONED","created":"2023-03-13 11:12:48.000000000","updated":"2023-03-13 11:27:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"b117c89a8443a3c1d3a98749b3dc11b0aead3f2e","_number":877252,"virtual_id_number":877252,"owner":{"_account_id":35849,"name":"Nik Reiman","display_name":"Nik Reiman","email":"nre@ableton.com","username":"nre-ableton"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-13 11:27:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-03-13 11:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb94f50fea9f1ee3fca6736fcc29347d1f4ef9e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35849,"name":"Nik Reiman","display_name":"Nik Reiman","email":"nre@ableton.com","username":"nre-ableton"},"date":"2023-03-13 11:12:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a32c3665ede1a9f1387ab135b067cb3ee539f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 11:27:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f966e3a46a3f432b9b02d66f62dbff9f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f9983776705e4169b6ae26d2abfd9206 : SUCCESS in 1m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1539df58634a65ad1e15356f83ee53 : RETRY_LIMIT in 1m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df79dc179d9c4dc38af8bc291afed006 : RETRY_LIMIT in 1m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c94150bfa8e4c7f897eed261b751c70 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ca4cd86592048878d597401c0e076de : SUCCESS in 4m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c6ce5edb6aa242ddb8888b491593a586 : RETRY_LIMIT in 1m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/382557ae70c648acb0d005723ab47cf0 : SUCCESS in 3m 58s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/e6103185f48c410d8f59aae4d787f588 : RETRY_LIMIT in 1m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b117c89a8443a3c1d3a98749b3dc11b0aead3f2e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35849,"name":"Nik Reiman","display_name":"Nik Reiman","email":"nre@ableton.com","username":"nre-ableton"},"date":"2023-03-13 11:27:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7adb1c02d43324e64c60edce8cd34e99e71cb0fa","revisions":{"7adb1c02d43324e64c60edce8cd34e99e71cb0fa":{"kind":"REWORK","_number":1,"created":"2023-03-13 11:12:48.000000000","uploader":{"_account_id":35849,"name":"Nik Reiman","display_name":"Nik Reiman","email":"nre@ableton.com","username":"nre-ableton"},"ref":"refs/changes/52/877252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/52/877252/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/52/877252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/52/877252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/52/877252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/52/877252/1"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Nik Reiman","email":"nre@ableton.com","date":"2023-03-13 11:12:48.000000000","tz":0},"committer":{"name":"Nik Reiman","email":"nre@ableton.com","date":"2023-03-13 11:12:48.000000000","tz":0},"subject":"Replace pkg_resources with packaging module","message":"Replace pkg_resources with packaging module\n\nChange-Id: I52aa4f3af07d6d14044fa54939ffea928051f953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/7adb1c02d43324e64c60edce8cd34e99e71cb0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/7adb1c02d43324e64c60edce8cd34e99e71cb0fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
