)]}'
{"id":"jjb%2Fpython-jenkins~877926","triplet_id":"jjb%2Fpython-jenkins~master~I6b7ea235cf0a3a6966c31d548586374502f1e4e3","project":"jjb/python-jenkins","branch":"master","topic":"hitesh","attention_set":{},"removed_from_attention_set":{"35874":{"account":{"_account_id":35874,"name":"Ted Chen","email":"ted.chen@barco.com"},"last_update":"2025-06-15 10:25:33.000000000","reason":"\u003cGERRIT_ACCOUNT_35866\u003e replied on the change","reason_account":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"}},"34567":{"account":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"last_update":"2025-06-15 10:50:55.000000000","reason":"\u003cGERRIT_ACCOUNT_34567\u003e replied on the change","reason_account":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}},"35866":{"account":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"last_update":"2025-06-15 14:44:15.000000000","reason":"Change was submitted"},"2475":{"account":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"last_update":"2025-06-15 14:30:17.000000000","reason":"\u003cGERRIT_ACCOUNT_35866\u003e replied on the change","reason_account":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"}}},"hashtags":[],"change_id":"I6b7ea235cf0a3a6966c31d548586374502f1e4e3","subject":"Added get_nodes_with_info() function to show information of all the nodes","status":"MERGED","created":"2023-03-21 07:22:13.000000000","updated":"2025-06-15 14:44:15.000000000","submitted":"2025-06-15 14:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877926-hitesh","meta_rev_id":"c662ef4c77e03ec6cd1ea76c7b531a31eda3d6f2","_number":877926,"virtual_id_number":877926,"owner":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":0,"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-15 14:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"all":[{"value":2,"date":"2025-06-15 10:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"tag":"autogenerated:gerrit:deleteReviewer","value":0,"date":"2025-06-15 14:30:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"all":[{"value":1,"date":"2025-06-15 14:30:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":0,"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 08:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 09:24:49.000000000","updated_by":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"reviewer":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"state":"CC"},{"updated":"2023-03-23 02:59:57.000000000","updated_by":{"_account_id":35874,"name":"Ted Chen","email":"ted.chen@barco.com"},"reviewer":{"_account_id":35874,"name":"Ted Chen","email":"ted.chen@barco.com"},"state":"CC"},{"updated":"2025-06-15 10:25:33.000000000","updated_by":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"reviewer":{"_account_id":35874,"name":"Ted Chen","email":"ted.chen@barco.com"},"state":"REMOVED"},{"updated":"2025-06-15 10:25:33.000000000","updated_by":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"reviewer":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"state":"REVIEWER"},{"updated":"2025-06-15 10:25:56.000000000","updated_by":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"reviewer":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"state":"REVIEWER"},{"updated":"2025-06-15 14:30:17.000000000","updated_by":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"reviewer":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"state":"REMOVED"}],"messages":[{"id":"b529d124a562f4ea8ffd20aaa2d3767636d1c142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 07:22:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"464403db1df55d3021a977168dc9419decef6a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 07:36:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"faf6016881a33f9111184df0a8fbf9735f8fe3f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 08:08:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ac43384d61a4e9c880849597135709d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f12f8cb1bc4c61ab2e88e0475c2f53 : RETRY_LIMIT in 1m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7241ad7d81894960828222bea424e314 : RETRY_LIMIT in 1m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/302806e53db645f398bdd0d26ae41f04 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88b53fb099c5409397f19c50bd848bed : SUCCESS in 3m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4254a674481f4412bc862a68c5c2b40b : RETRY_LIMIT in 1m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2204345020d74486b29ddf7601726720 : SUCCESS in 3m 31s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/c96e00a0a68942098f2ae7fc2a751ae9 : RETRY_LIMIT in 1m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b9acace4272c2cdc177b4878513fe84cfaaa1e6c","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 09:14:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7608e11a981a156376163f530ff100796f17503","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2023-03-21 09:24:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24db5968b06eb427694fdeea7fb5704445026b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 09:43:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98234c0a6e1b43a09ae8919b69b47391\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57cf0df3d8e34b29a83c4bfc853c1c44 : RETRY_LIMIT in 2m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30a752eb2cbe4acfb828bd7367353378 : RETRY_LIMIT in 1m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fa457688643040a1b380a5218e6bde21 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cc9ffe011794ae0802c4868ed60dd26 : SUCCESS in 4m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/551e55df788749e4b3087b816fb05745 : RETRY_LIMIT in 1m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/004c998b3683471fb388cf71ff1290d8 : SUCCESS in 3m 40s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/f044349d19b04a3cbb67e636e225361a : RETRY_LIMIT in 1m 39s","accounts_in_message":[],"_revision_number":2},{"id":"972f3b5d054cfdd731545245593f0aae825f6f4f","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2023-03-21 10:37:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e02fea02c82fca78fa0122e1849bc42c5ea4523f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 10:49:27.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8f19d8254e449c890cf07f2ff0126d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a132e1d423ff47a9a995f1e6265381f8 : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04bdb9bb9064417f992d45a543728231 : SUCCESS in 6m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c56debe1f924caab57914af0d117f8d : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cf4cc84c4c046ff830a201d2089d874 : SUCCESS in 5m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2f3b972fe90414b96f14243114513a1 : FAILURE in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79bd6526f2f64ac0876c9372bef0d66f : SUCCESS in 5m 59s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/54e9479f712a4ac18ba6088ea380182f : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c4778d4f4ee8a02179708505bba949c99934cd67","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 11:00:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef022817e5119fe28837125d9efc95b2d074eeab","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2023-03-21 11:06:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35188c8ce3f2d89479c70c34f711bf51eec6d974","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 11:15:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb94d9f9b5a2cfcd652e5ad03026ed87c6ca70cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 12:01:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67cdb5b01a5cd9254d8acf46e05e4976702b3e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 12:15:51.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e904fb63cc2a4899aafe83dfce9b7c6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a66ff47628461fa74b754b838ce86d : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2097f1bd5a54722a6bbf59145dd6497 : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c874347d08a44eb8a1fd328618f0cfad : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5453de37a96441f6acafdde0a18c1b84 : SUCCESS in 4m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e8c5fae06f74d5594d1a99765f411db : FAILURE in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97825eb60b0143709af828017e0bec86 : SUCCESS in 4m 06s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/b0a5fb1265f94022a164ff0ff7ae22b9 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"e2b1809668d23dd708b92f0c5d6580ac20b76b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-21 12:21:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e038628018095c5d08e40b35b3cd00edc0deb53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 12:31:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebb4adcc8c574491a3b8899c765ad56d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4394977be5247d48de464bbe0091982 : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4841dcd02954bb89a755633235a0285 : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67be9641a5ce4ed5a7404d57f8491b83 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9b06ccc2654239a17b161ae14925e1 : SUCCESS in 4m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a60f988ebda4e57b45daa00deb7e839 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1723643e0074c329f87657feef74534 : SUCCESS in 3m 46s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/e747bb611511434fa686a9478ef430bd : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ac328e366b4f51574fca2a7a143a0f4748ed7c6c","author":{"_account_id":35874,"name":"Ted Chen","email":"ted.chen@barco.com"},"date":"2023-03-23 02:59:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"13b37cb21b9420e31501ae2be68709838078d44d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-24 06:03:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a25d9e23f79e6059ebf8a56bae0393ae3a5ac0ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-24 06:08:06.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"61f7927f68b30b60a25ee15162dcceee9e6840a3","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-24 06:15:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"09bee6b3169278cc21e1e8a6490777e85f4a5e5e","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-24 06:17:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4e595cdb568f753ff2148dc6186ee37c06f0ff90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 06:20:04.000000000","message":"Patch Set 6: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d797ad05404a45d497c8a3800b83a9a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47beacf46c954d8caf10b30d628f7511 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cd79498337c431289ecedbcf68a61ef : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/467d460e4f71484dba3e45b8c6c7d6b1 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fa49edcad074c6a87ffa4943ab436f2 : SUCCESS in 3m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6121d235ccb14a2288a50df8a5bfff44 : FAILURE in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c87bc08d7ef427ebc1f30ea88e6f88c : SUCCESS in 3m 32s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/98bd4b51ec7949ea941e7a8e7e055cfb : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":6},{"id":"f51e41745d44f5ac9c325b17f2276fdfd19776d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2023-03-24 06:23:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e3f9f3f36ee93038327890dc4d8f0bbb9792c000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 06:31:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972e1dd134ab4ff89351c5d3e59e0e3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1938c213f7c14e15ac0d38834f4a6f11 : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8008b8ba864c415da611f5e9004d7345 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c8396305531247b19e70f4e5fa69d30d : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c3dfc9b9246403ea5e9561db599ec23 : SUCCESS in 3m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d57db96b0bd44c1da8540d84536d581c : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d229f8d135b407a90cc79223e54041d : SUCCESS in 3m 58s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/53afce6f2beb426c8f49585e7ebb119b : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":7},{"id":"7ad3e5897a90973b76a0de696d8cfa3ac2ae85ac","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2025-06-15 10:00:11.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fbd4c1c877b58f84cfb0202ac37986debba37b45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2025-06-15 10:00:47.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"4037088e51da03ed30280b935848e76fecbb53d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 10:12:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d40afa8189545d5bc4b69bcc6fc0799\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/16e0f5a2e43c4cc5ae9ad691a87a5c0c : SUCCESS in 6m 15s\n- tox-linters https://zuul.opendev.org/t/openstack/build/a519572a85344077835c82061ff3bd69 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5759676b6a4ac4a107671179eea0b8 : SUCCESS in 5m 56s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/2fe152dd195a4a698168fa9edbd6fb85 : SUCCESS in 3m 16s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/d13021d7f4534b82a5f54370ad75523e : SUCCESS in 5m 55s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/bd2373bd11554a45889d1807bb2f0580 : SUCCESS in 3m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cd27cc6fbeb24a168fcf5437fe86007c : SUCCESS in 3m 14s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/b5ea64cee4194f819f8d431e4501b0bc : SUCCESS in 7m 36s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/167842857042436985794880b87a1bf7 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":8},{"id":"944da1f4a37332c8ef7da0408bc276ee6f31edba","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2025-06-15 10:25:33.000000000","message":"Patch Set 8: -Code-Review","accounts_in_message":[],"_revision_number":8},{"id":"e1de2dca59bd6db86a64ab60f369c24745a9a819","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-06-15 10:50:55.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3ef3e1556a42ee6275d7ebf798accbc72feec635","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2025-06-15 14:21:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3311defed0338d16969d4957097bcbf044783b45","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-06-15 14:23:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7a8bf0094568beb1b094546dfb9ae57108991c09","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2025-06-15 14:29:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"51a739e21d895ac3a3b3785332bb89297134e154","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"date":"2025-06-15 14:30:17.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_2475\u003e.","accounts_in_message":[{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"}],"_revision_number":8},{"id":"f960ea87780552a6e8ebbd593fe49f194ff7dc25","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-06-15 14:30:39.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ea535308be4e2ea0c56b5e3baddaa494bb70766a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 14:30:48.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b7066ad7870bb576b985aeca572c288ce45a0473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 14:44:15.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57631f7b786447e18d77d52e3085fba8\n\n- tox-linters https://zuul.opendev.org/t/openstack/build/ea4b02d1cd054807aae62cff92ca3d1f : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf66672fcc8744309944c7e0b1115350 : SUCCESS in 3m 27s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/c4231d9b5c544f16a1d1d8574ba16cf5 : SUCCESS in 3m 13s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/dc7873c7d4e04d79aba2036dad31d845 : SUCCESS in 3m 21s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/d22833b2570a480d95d5f6452dd568a3 : SUCCESS in 3m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b90d4d86e1bf4ad7a1d91d863e609f22 : SUCCESS in 5m 32s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/7dfeb1d8cbbf4dab85564b6a56a03a54 : SUCCESS in 6m 16s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/64c7715b618e4574a4c8f781cd69faf5 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":8},{"id":"c662ef4c77e03ec6cd1ea76c7b531a31eda3d6f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 14:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6c2e3be09a78d1b5f9be894188b5b347086bd078","revisions":{"0fca141b6925f8bdeccbf177fc457192984834e8":{"kind":"REWORK","_number":1,"created":"2023-03-21 07:22:13.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/1"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/0fca141b6925f8bdeccbf177fc457192984834e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/0fca141b6925f8bdeccbf177fc457192984834e8"}]},"branch":"refs/heads/master"},"f86c9161209f1a13cf91a33f706b03734b76f100":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-21 07:36:35.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/2"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-21 07:36:35.000000000","tz":0},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nThis function will reduce load time since we will not need to call the get_node_info() for individual nodes and can get information on all the nodes in one call.\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/f86c9161209f1a13cf91a33f706b03734b76f100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/f86c9161209f1a13cf91a33f706b03734b76f100"}]},"branch":"refs/heads/master","description":"Edit commit message"},"792b07fe11acfc7e607358f000cfdc5155ae8016":{"kind":"REWORK","_number":3,"created":"2023-03-21 12:01:50.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/3","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/3"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-21 11:59:22.000000000","tz":330},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/792b07fe11acfc7e607358f000cfdc5155ae8016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/792b07fe11acfc7e607358f000cfdc5155ae8016"}]},"branch":"refs/heads/master"},"bb772c3497d4b7891060993c7acadd15dcfdb5f5":{"kind":"REWORK","_number":4,"created":"2023-03-21 12:21:55.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/4","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/4"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-21 12:20:20.000000000","tz":330},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/bb772c3497d4b7891060993c7acadd15dcfdb5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/bb772c3497d4b7891060993c7acadd15dcfdb5f5"}]},"branch":"refs/heads/master"},"e607651ad8578ddd40e22ef2e457184d42c568d4":{"kind":"REWORK","_number":5,"created":"2023-03-24 06:03:19.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/5","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/5"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-24 06:00:38.000000000","tz":330},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/e607651ad8578ddd40e22ef2e457184d42c568d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/e607651ad8578ddd40e22ef2e457184d42c568d4"}]},"branch":"refs/heads/master"},"5eb8afaf9b1ff61107a4eb01b7160dd5271accbb":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-03-24 06:08:06.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/6","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/6"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-24 06:08:06.000000000","tz":0},"subject":"Added get_nodes_with_info() function to show information on all the nodes","message":"Added get_nodes_with_info() function to show information on all the nodes\n\nThis function will fetch the information of all the nodes at once, avoiding the overhead of calling the get_node_info() for all the individual nodes and hence, reducing the time to fetch data of all the nodes.\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5eb8afaf9b1ff61107a4eb01b7160dd5271accbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5eb8afaf9b1ff61107a4eb01b7160dd5271accbb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8172edd32c6117b9957578a9cd9b75dc24e2f980":{"kind":"REWORK","_number":7,"created":"2023-03-24 06:23:28.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/7","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/7"}}},"commit":{"parents":[{"commit":"d1e469649574883bc531d35fb3aee207f45c15a5","subject":"Handle new master naming in Jenkins post v2.307.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/d1e469649574883bc531d35fb3aee207f45c15a5"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-24 06:21:59.000000000","tz":330},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nThis function will fetch the information of all the nodes at once,\navoiding the overhead of calling the get_node_info() for all the\nindividual nodes and hence, reducing the time to fetch data of all the\nnodes.\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/8172edd32c6117b9957578a9cd9b75dc24e2f980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/8172edd32c6117b9957578a9cd9b75dc24e2f980"}]},"branch":"refs/heads/master"},"6c2e3be09a78d1b5f9be894188b5b347086bd078":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-06-15 10:00:47.000000000","uploader":{"_account_id":35866,"name":"Hitesh Malhotra","display_name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","username":"hitesh1409"},"ref":"refs/changes/26/877926/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/26/877926/8","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/26/877926/8"}}},"commit":{"parents":[{"commit":"8af7e92d4714e256a75a7d86eac2551554dc75ab","subject":"Add tree parameter to get_build_test_report()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/8af7e92d4714e256a75a7d86eac2551554dc75ab"}]}],"author":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2023-03-20 08:37:50.000000000","tz":330},"committer":{"name":"Hitesh Malhotra","email":"hiteshmalhotra622@gmail.com","date":"2025-06-15 10:00:47.000000000","tz":0},"subject":"Added get_nodes_with_info() function to show information of all the nodes","message":"Added get_nodes_with_info() function to show information of all the nodes\n\nThis function will fetch the information of all the nodes at once,\navoiding the overhead of calling the get_node_info() for all the\nindividual nodes and hence, reducing the time to fetch data of all the\nnodes.\n\nChange-Id: I6b7ea235cf0a3a6966c31d548586374502f1e4e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/6c2e3be09a78d1b5f9be894188b5b347086bd078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/6c2e3be09a78d1b5f9be894188b5b347086bd078"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
