)]}'
{"id":"jjb%2Fpython-jenkins~888285","triplet_id":"jjb%2Fpython-jenkins~master~I6b5327a5f969607ca4838a13bbac838f44e76ccb","project":"jjb/python-jenkins","branch":"master","topic":"view-all-name","attention_set":{},"removed_from_attention_set":{"34567":{"account":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"last_update":"2023-08-19 10:07:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34567\u003e replied on the change","reason_account":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}},"10089":{"account":{"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},"last_update":"2023-08-19 10:32:47.000000000","reason":"Change was submitted"},"31275":{"account":{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},"last_update":"2023-08-19 10:32:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b5327a5f969607ca4838a13bbac838f44e76ccb","subject":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027","status":"MERGED","created":"2023-07-12 14:49:20.000000000","updated":"2023-08-19 10:32:47.000000000","submitted":"2023-08-19 10:32:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888285-view-all-name","meta_rev_id":"19562b3507e1e2693bbbb53dd57b9b1113d465cb","_number":888285,"virtual_id_number":888285,"owner":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":0,"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},{"value":0,"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-19 10:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"all":[{"value":2,"date":"2023-08-19 10:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":1,"date":"2023-08-19 10:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},{"value":1,"date":"2023-08-19 10:06:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"all":[{"value":1,"date":"2023-08-19 10:07:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":0,"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},{"value":0,"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 14:50:36.000000000","updated_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"reviewer":{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},"state":"REVIEWER"},{"updated":"2023-07-12 14:50:36.000000000","updated_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"reviewer":{"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},"state":"REVIEWER"},{"updated":"2023-07-12 15:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5bb88f3b9383a3db3b101c530599227eb85ddfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-07-12 14:49:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"136fe1131e0f81a842238dd71b4c510efddd4724","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-07-12 14:50:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3c2055842c840815032125b3564c6b39ad606a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 15:23:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80a378d05db14d3ea2abe82dcdf7b20f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d504060d14c4f87872156760cde5540 : FAILURE in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bac51cf451546a1824b5f498f2c3a16 : FAILURE in 3m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b16e5731a5a48119dbc5055761a6eaf : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b52b407428c6450bbf8cba9343e41968 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b2d0a016bba4f9e8ee4ae3889d401ea : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e65da56c06104ada9423a82833fdbfe1 : SUCCESS in 4m 37s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/22b9a18c32224378be2a54dc3065ad9d : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9b5566c30b13fbee9fe4041ff6a2d54f5662e6b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-07-17 12:12:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"feb1e68a7ef78ecc4c3ca6492d67970be5c2acc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:36:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970645395b4b4ac99ac2c77d6404b6d2\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/46be1893970a4f7b988ef4f706e74c50 : SUCCESS in 3m 49s\n- tox-linters https://zuul.opendev.org/t/openstack/build/c9f63b50c9194342923bcf1b4cd7e471 : SUCCESS in 2m 37s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a47ebd4095304636ae39bb4b6aff9413 : SUCCESS in 4m 14s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/37a71aef391b456ca1efd235e68d7697 : SUCCESS in 4m 20s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/0d67e1ce69814cf48983aaead9b71d11 : SUCCESS in 4m 38s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/d27cb556d6474316b361a05c5707c3dd : SUCCESS in 3m 23s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/8e9f5b3cfada45acb0fcf9353e690461 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3d35ab25824216b93338f45cb834b6a02f2d5d36","author":{"_account_id":10089,"name":"Anil Shashikumar Belur","email":"abelur@linuxfoundation.org","username":"askb"},"date":"2023-07-18 04:11:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"93ff9f54e25108ba3cb377c1b0aaab3b64ffc430","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-07-18 06:39:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c104648e9e77c2c0dcc2bb198f507ce805f2afc","author":{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},"date":"2023-07-18 10:53:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b2e33738334ec5e3f1a160295630f2c35d672327","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-08-19 10:06:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"61eb91d49d1011a9e986a45f96d48225639c452d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-08-19 10:06:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"6c79275f500c4ac4f90cbb370e9b242bc9f25062","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2023-08-19 10:07:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6dc81c798a12a457fea20cae9fe91cb44b3578b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 10:20:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b52831e971fa45afa655b17cd481f168\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/aef83903dd00431290ad30875b8c9457 : SUCCESS in 4m 00s\n- tox-linters https://zuul.opendev.org/t/openstack/build/30bf264828134bc28654c897b9d716b1 : SUCCESS in 2m 45s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/58accb308d9044c297f75a5bdc34b63f : SUCCESS in 3m 14s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/59627c7550af4fabae4fe3b33d54641b : SUCCESS in 4m 25s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/39e67135f85f483e89200bc5f808f9a2 : SUCCESS in 3m 46s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/37189e91a2f04588b936f2bd3ac069da : SUCCESS in 4m 09s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/e282055f37aa444c9d04022bead9ff6a : SUCCESS in 6m 03s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/915bf9eb89154b92a923c1770aa98aa2 : SUCCESS in 3m 11s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/ed93d6d00e8c4ff5be0192cf9499cf5d : SUCCESS in 6m 25s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/fe44fddfac034573b8c1eda1dd942805 : SUCCESS in 3m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ca2c558df13242be941620df428b152d : SUCCESS in 5m 54s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/e972f465000445acb5148c131fd0c6c3 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"724c0c4702ad0e9d1207a5ede5b03079e18224f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 10:20:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7bf76892a78aa659715f5450f88523bb1a1c1b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 10:32:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6b7ba0b7ef342ac9752446c0839252e\n\n- tox-linters https://zuul.opendev.org/t/openstack/build/aae485d88cd543419dc87e051d444bad : SUCCESS in 2m 35s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/dd2a25f657874d45ba720641285450fd : SUCCESS in 3m 11s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/e4f175be63ea4200bcf118a147c694b0 : SUCCESS in 3m 10s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/058fe74fdafa403f832224ec1dbe5961 : SUCCESS in 3m 38s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/dfe6a688a2264a78a79b0b5d354ee3e8 : SUCCESS in 3m 00s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/79e1a1d45eb04d9eb2a0c72e9759e8cc : SUCCESS in 4m 48s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/db690090f1a94a91a901ef36cae6023c : SUCCESS in 2m 49s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/5f216b4a58b1425192d44fb0b9f993af : SUCCESS in 4m 55s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/fcf171a339ce4680b22dc94bd319b882 : SUCCESS in 2m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3e2482ef3cf944d3998ef05a63e0610f : SUCCESS in 4m 27s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/c9af6645a3a04d22a3f203160410ae14 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"19562b3507e1e2693bbbb53dd57b9b1113d465cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 10:32:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a8f0d8cfda832c700cf4ca744d07acff3afc5f04","revisions":{"1cf137d392dce73acb811186b5930cc63a2add2c":{"kind":"REWORK","_number":1,"created":"2023-07-12 14:49:20.000000000","uploader":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"ref":"refs/changes/85/888285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/85/888285/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/1"}}},"commit":{"parents":[{"commit":"70dc3e4f711088e0cc9f0a92706fee3cb0757bac","subject":"Fix CI tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/70dc3e4f711088e0cc9f0a92706fee3cb0757bac"}]}],"author":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2023-07-12 14:46:11.000000000","tz":180},"committer":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2023-07-12 14:49:18.000000000","tz":180},"subject":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027","message":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027\n\nJenkins version 2.387.1 or earlier can return \u0027all\u0027 as view name\nwhen requested is \u0027All\u0027. Add workaround for that.\n\nChange-Id: I6b5327a5f969607ca4838a13bbac838f44e76ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/1cf137d392dce73acb811186b5930cc63a2add2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/1cf137d392dce73acb811186b5930cc63a2add2c"}]},"branch":"refs/heads/master"},"a5f7431e063e002d9269d72f017f6f9118eff307":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-17 12:12:57.000000000","uploader":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"ref":"refs/changes/85/888285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/85/888285/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/2"}}},"commit":{"parents":[{"commit":"628c3d00e9e6fdba3d822a68f919f3b5fc621a37","subject":"CI: move out of OpenStack Zuul templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/628c3d00e9e6fdba3d822a68f919f3b5fc621a37"}]}],"author":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2023-07-12 14:46:11.000000000","tz":180},"committer":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2023-07-17 12:12:57.000000000","tz":0},"subject":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027","message":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027\n\nJenkins version 2.387.1 or earlier can return \u0027all\u0027 as view name\nwhen requested is \u0027All\u0027. Add workaround for that.\n\nChange-Id: I6b5327a5f969607ca4838a13bbac838f44e76ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a5f7431e063e002d9269d72f017f6f9118eff307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a5f7431e063e002d9269d72f017f6f9118eff307"}]},"branch":"refs/heads/master","description":"Rebase"},"a8f0d8cfda832c700cf4ca744d07acff3afc5f04":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-19 10:06:25.000000000","uploader":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"ref":"refs/changes/85/888285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/85/888285/3","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/85/888285/3"}}},"commit":{"parents":[{"commit":"5a98031aeed375cd570e487ebfb34edc67f25a5f","subject":"WrappedSession: Fix URL to requests issue 3829","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/5a98031aeed375cd570e487ebfb34edc67f25a5f"}]}],"author":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2023-07-12 14:46:11.000000000","tz":180},"committer":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2023-08-19 10:06:25.000000000","tz":0},"subject":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027","message":"STORY 2010790: Handle \u0027all\u0027 view name when requested \u0027All\u0027\n\nJenkins version 2.387.1 or earlier can return \u0027all\u0027 as view name\nwhen requested is \u0027All\u0027. Add workaround for that.\n\nChange-Id: I6b5327a5f969607ca4838a13bbac838f44e76ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a8f0d8cfda832c700cf4ca744d07acff3afc5f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/a8f0d8cfda832c700cf4ca744d07acff3afc5f04"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
