)]}'
{"id":"jjb%2Fpython-jenkins~944779","triplet_id":"jjb%2Fpython-jenkins~master~I9483343208d6f1be7b2b74ba993499c51852f431","project":"jjb/python-jenkins","branch":"master","attention_set":{},"removed_from_attention_set":{"34567":{"account":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"last_update":"2025-04-15 18:40:01.000000000","reason":"\u003cGERRIT_ACCOUNT_34567\u003e replied on the change","reason_account":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}},"31275":{"account":{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},"last_update":"2025-04-15 19:06:39.000000000","reason":"Change was submitted"},"37838":{"account":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"last_update":"2025-04-15 19:06:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9483343208d6f1be7b2b74ba993499c51852f431","subject":"Add support for connection retry","status":"MERGED","created":"2025-03-17 14:21:19.000000000","updated":"2025-04-15 19:06:39.000000000","submitted":"2025-04-15 19:06:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944779","meta_rev_id":"ca01cfe4ceb78ed9e01bc2acef63018b86206d28","_number":944779,"virtual_id_number":944779,"owner":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-15 19:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"all":[{"value":2,"date":"2025-04-15 18:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"all":[{"value":1,"date":"2025-04-15 18:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-17 14:27:03.000000000","updated_by":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"reviewer":{"_account_id":31275,"name":"Artem Nikitin","email":"anikitin@networkoptix.com","username":"artem_nikitin"},"state":"REVIEWER"},{"updated":"2025-03-17 14:27:03.000000000","updated_by":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"reviewer":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"state":"REVIEWER"},{"updated":"2025-03-17 15:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db93bcaea76e71d604244f30af9eeca951780900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"date":"2025-03-17 14:21:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a47ea4829fab623ac9a6869e811053ca9b88a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-17 14:34:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"193370617ff49875aefd64e2cb2ad69a841c30ec","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-17 14:34:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad605530674e0574cca0f6e4c0cde646286e45f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 15:19:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd39608c1f48401195ffb0428380fae3\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/284efd58409d408bbe16d41190bd4250 : RETRY_LIMIT in 6s\n- tox-linters https://zuul.opendev.org/t/openstack/build/617241c1422b429c9a92daacd8a50745 : RETRY_LIMIT in 1m 30s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/026584e66ffc48d8903b8b5efd85cb87 : RETRY_LIMIT in 7s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/460b5712dabd4b6fa811e77a8343cc11 : RETRY_LIMIT in 6s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/71f24195095644a8ac1f41214d8e6d9a : RETRY_LIMIT in 6s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/a9ca54395ead409b9004e8bdd2ab5d7f : FAILURE in 3m 29s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/d0fe778d674544efaba8f358fe01538d : FAILURE in 7m 51s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/b08e6ebed0b74ccaa654017421020c2e : FAILURE in 5m 51s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/a97d3fcf3e564603ba0bd5a600abad68 : FAILURE in 8m 32s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/3720756ce513472b852b258d31f5bdaa : FAILURE in 5m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/2551319bb2fe47989af712ba4c6cee42 : FAILURE in 7m 38s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/03fbef951e884c1c8c40ed7285344cc5 : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f947372fa2c7ff88a6f1453b35bfcdf927df8185","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-17 15:31:27.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"8a1d1b55dab43765e93d4c6b90d9d03ede2f8b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"date":"2025-03-18 06:22:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"05330393edf40a372e5a9f2fa0c70ef6c1f4a3e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 06:49:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed4e32f17062424eaa543f7ec55c2f51\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/fc73d4f765cc41b1bd4620baac56f263 : RETRY_LIMIT in 6s\n- tox-linters https://zuul.opendev.org/t/openstack/build/d79adab031f14640bad9721f9bec2a54 : RETRY_LIMIT in 3m 54s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f5f325e05a0c47b0b3646f4be0a5a806 : RETRY_LIMIT in 7s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/077236c493c9489888d1ee1fcfeb8c3c : RETRY_LIMIT in 6s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/16ebc5989a6a49d3bf70a400cb377142 : RETRY_LIMIT in 7s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/849c78cc5a984f1d9a786ccd60f530e9 : SUCCESS in 6m 36s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/67bf1b17b6dc47e7b9bc0bf11407f32f : SUCCESS in 5m 32s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/b9692293eceb43c2a51907abe98f15e2 : SUCCESS in 8m 35s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/4f7aa3ae31494bf0997b5acd49bdfc03 : SUCCESS in 4m 58s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/2c7fc27f55de4477a71964e15f6d0b56 : SUCCESS in 2m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e36e1ce2f8ca4ca3bdcab6ff05c55ca1 : SUCCESS in 8m 06s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/fd8387c5351b4f6989061afa9e934415 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":3},{"id":"cdd9a01e64b3f5aad49f5abf737993b3dd7ce50b","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"date":"2025-03-19 14:24:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa42890c9768260d56b6c676557d5d1000118e41","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-19 14:44:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"811ff492f9a2da16926e0bb51693a3e92a7a2b9f","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-19 14:44:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ba31c9b5d1ebb5e254d66f5b43673e0cfebfee6","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-19 14:50:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac855c2ef9969542faf2958bc82c61a7fbec4918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 15:19:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a13d43f144104217b3a68124f85f6fa0\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/d9a8c7a2667540b6a676852925798eb3 : RETRY_LIMIT in 6s\n- tox-linters https://zuul.opendev.org/t/openstack/build/1b777e284704495385f20532cd48b3b3 : RETRY_LIMIT in 1m 37s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c98e2a8ffc8041b699e9fc23fb608ffe : RETRY_LIMIT in 6s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/dc28bba65d3548b8ba05e888ceea3259 : RETRY_LIMIT in 7s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/e6c28b7323d34846883f82688730f183 : RETRY_LIMIT in 7s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/39dd7b2224ad426daca3a2ab5d978de5 : SUCCESS in 3m 58s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/984845baf9f94490b3091f7f6aefc512 : SUCCESS in 8m 24s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/d79be39e07d64f0faaf6c61d6271d7f9 : SUCCESS in 7m 06s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/480a163cf4114c98979f1a6d07ae7c8d : SUCCESS in 10m 42s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/f4e53814a8204dbfaa717385dcb64279 : SUCCESS in 5m 38s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0cc64d5983b3424c9405947d59de869a : SUCCESS in 3m 47s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/67e7369e89b24c55956fd7736989fc3b : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":3},{"id":"03984ee0fca88ebee50d5a7141aa5e5e34bb6a13","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-19 15:32:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"929e25f3ae7ea0664bc4efd27a05a98dc813a736","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"date":"2025-03-20 09:36:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4400ff1a81dd904d272832a12acccce4a8d68657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 10:27:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0761249c92d4ad299a02b9fc3cda51a\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/219e4afb7fa14299bc941557212e86bc : RETRY_LIMIT in 6s\n- tox-linters https://zuul.opendev.org/t/openstack/build/073edf2219a441ba8793f56054fdf926 : RETRY_LIMIT in 1m 27s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/899894cd36584a129d2aa86641c54acc : RETRY_LIMIT in 5s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/8cc5f8836abf4d20a995093dd73379fe : RETRY_LIMIT in 6s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/5208163c285a44e4b5159feef0e1bc19 : RETRY_LIMIT in 6s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/035b1f4e0cd44eeaa2f02dbc6d18d327 : SUCCESS in 6m 07s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/b65ae4db33a34ef9adbea069bce7057f : SUCCESS in 6m 46s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/658ede4a6edd429aa32e70d8e73af643 : SUCCESS in 5m 09s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/149f340b7de3475f8dfdd389afb52e9d : SUCCESS in 3m 58s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/38576b1b390c4f30a78e6f7ee202f0c0 : SUCCESS in 4m 56s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3ca88f1a4370423aa10914c50e77c832 : SUCCESS in 3m 38s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/01d87396ad3742b4a7a982080098426a : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ebec62c1db9a823d156ac3dfb908c92867351c83","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-03-20 10:52:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3c4d5d0eb2e09715773b4b3c23b6e02951ec8d7c","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"date":"2025-03-25 10:04:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f5fe82f2505fabf37383afa6c94e289049961e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 10:48:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d03ffeb66caf49138c4b9f0c2db238e3\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/6303d233dcca4da0b0a0d22be3b6b903 : RETRY_LIMIT in 6s\n- tox-linters https://zuul.opendev.org/t/openstack/build/f21e32310b1d48879838077aa06c9bbc : RETRY_LIMIT in 1m 48s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/40034e9e192d4c7989da0486391ac22f : RETRY_LIMIT in 6s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/db3a92087d69411a8b73773e5b9dda25 : RETRY_LIMIT in 8s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/615c54c60ee349e4823934e54e619abc : RETRY_LIMIT in 7s\n- tox-py36 https://zuul.opendev.org/t/openstack/build/0081ce08cf7340bbab2b0452561d2393 : SUCCESS in 6m 11s\n- tox-py37 https://zuul.opendev.org/t/openstack/build/55671ac160cd47b4a8ef4857cb66f2cb : SUCCESS in 7m 31s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/58423fe0cee347efada038e0b81ce3ac : SUCCESS in 6m 08s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/4e63f647311d49809ff8f9bb4db7360c : SUCCESS in 7m 31s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/bfc7828e8ede4f22bcbf55110b5c6ff7 : SUCCESS in 15m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bae0eab504ce4b22a807f5af9f1fd9b6 : SUCCESS in 6m 56s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/af5bc391f1844706a4c50906bd9d8968 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5d2adbbeaed02b090f21666b2d461addb22cdfae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"real_author":{"_account_id":34829,"name":"guillaume de mengin","display_name":"gdemengin","email":"gdemengin@gmail.com","username":"gdemengin"},"date":"2025-04-12 17:52:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_37838\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"}],"_revision_number":4},{"id":"89917f0d80d6be87f89ad14612dfc638a8d5e41b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 17:58:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c801c2b9bdf41358c6ebae7e54c7afa\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/f4e9c4e04dda4805965c3b6be1b18326 : SUCCESS in 4m 57s\n- tox-linters https://zuul.opendev.org/t/openstack/build/c49e834c3f634de29ca5d3eb7132af6a : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1a4384b9c341a68fea892733d7c1b7 : SUCCESS in 2m 39s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/bee42a78a9e047da997202c70250d50b : SUCCESS in 2m 57s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/9cb617a3033c4cac8e63bc25e10b466c : SUCCESS in 3m 32s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/36ee7771fe1b4ebea01df5603163f1ff : SUCCESS in 2m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e2b05236673f4da9b03802ce68366873 : SUCCESS in 4m 11s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/d1a4a01b43854ca08a4b1ebbee8b95c3 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"6cd8831808473a16d5594eb7e0db65efd32a1653","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-04-12 18:00:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8b504521d8b3f37b4df247d1c9af399938548359","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-04-14 16:44:07.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5223a8f7a4bdb140337ec2782e17112ee35636b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"real_author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-04-15 18:39:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_37838\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"}],"_revision_number":5},{"id":"d0024106f6754f0c18ea801302c0dc19bf09cef4","author":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"date":"2025-04-15 18:40:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"03064ac98640b40cf467ce2e9d538bdd663e5d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:53:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2857caa1a2af44d9963752e119f2b0ac\n\n- tox-cover https://zuul.opendev.org/t/openstack/build/546b1aa1361b4b04907d08290dfe4684 : SUCCESS in 8m 34s\n- tox-linters https://zuul.opendev.org/t/openstack/build/b45a6a2332f647b4892155ce6cd4acb6 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0429aa491cd243bc9eea7153b8ed0961 : SUCCESS in 8m 03s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/4f9c365e48db47fa89d34b6f7bb6c37c : SUCCESS in 5m 02s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/2999348c5a99437eacfab22a19655705 : SUCCESS in 8m 05s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/5ae4b4799f014391a86a4ca22ba632ff : SUCCESS in 5m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/4343e793ec6b40f984b42af2b06fc312 : SUCCESS in 7m 16s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/99165b937c0349199619d1c5f4a1bbd8 : SUCCESS in 8m 59s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/183b08a539804ea3862e05e91246d57d : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":5},{"id":"10b2780d2dbe46edb8797a5cfb062bf2e4640623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:53:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a65ad05a54b7eee5af90b1115ab81e018ea2f19e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 19:06:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ed333ab6ec4ce0bdcf5c9dfaa9218f\n\n- tox-linters https://zuul.opendev.org/t/openstack/build/51d3731106f741a6a380425e437d91d5 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ddf14f66a8430ebc1c9cd204ffc925 : SUCCESS in 8m 22s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/c3e50c6af58843209df1a4511459b120 : SUCCESS in 6m 11s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/725fcf8b832b46a3b8f63adbbdfbab1c : SUCCESS in 8m 30s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/aeccc9c47d8a47808457311ee057c5fc : SUCCESS in 5m 21s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/378d06397e9b47338b21050a2c07daf4 : SUCCESS in 7m 28s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/ae36a30f127d4997b72fa8c34b90a9f2 : SUCCESS in 8m 11s\n- jjb-tox-cross-jenkins-job-builder https://zuul.opendev.org/t/openstack/build/90edc717d62f47379d42fe31efd6361e : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ca01cfe4ceb78ed9e01bc2acef63018b86206d28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 19:06:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f29d64f991b7d44140863e0f4efa17633793e949","revisions":{"daae339f5b002b38a8229ff2dfdcad7f08409985":{"kind":"REWORK","_number":1,"created":"2025-03-17 14:21:19.000000000","uploader":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"ref":"refs/changes/79/944779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/79/944779/1","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/1"}}},"commit":{"parents":[{"commit":"b147d864468614dd3e2e64304c71847c069f293b","subject":"only change view name from all to All when All was requested","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/b147d864468614dd3e2e64304c71847c069f293b"}]}],"author":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-17 14:08:28.000000000","tz":60},"committer":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-17 14:11:24.000000000","tz":60},"subject":"Add support for connection retry","message":"Add support for connection retry\n\nWith the new optional parameter `retries` the number of retries in case of a\nfailure can be specified.\n\nFor compability reasons the default is kept as zero.\n\nChange-Id: I9483343208d6f1be7b2b74ba993499c51852f431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/daae339f5b002b38a8229ff2dfdcad7f08409985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/daae339f5b002b38a8229ff2dfdcad7f08409985"}]},"branch":"refs/heads/master"},"dbddbb5b8ce5c7aaaea89d026c96242402013472":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-17 14:34:15.000000000","uploader":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"ref":"refs/changes/79/944779/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/79/944779/2","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/2"}}},"commit":{"parents":[{"commit":"b147d864468614dd3e2e64304c71847c069f293b","subject":"only change view name from all to All when All was requested","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/b147d864468614dd3e2e64304c71847c069f293b"}]}],"author":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-17 14:08:28.000000000","tz":60},"committer":{"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","date":"2025-03-17 14:34:15.000000000","tz":0},"subject":"Add support for connection retry","message":"Add support for connection retry\n\nWith the new optional parameter `retries` the number of retries in case\nof a failure can be specified.\n\nFor compability reasons the default is kept as zero.\n\nChange-Id: I9483343208d6f1be7b2b74ba993499c51852f431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/dbddbb5b8ce5c7aaaea89d026c96242402013472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/dbddbb5b8ce5c7aaaea89d026c96242402013472"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3baa5b2c3c950790899323579f4fd2c0ea85cd86":{"kind":"REWORK","_number":3,"created":"2025-03-18 06:22:49.000000000","uploader":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"ref":"refs/changes/79/944779/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/79/944779/3","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/3"}}},"commit":{"parents":[{"commit":"b147d864468614dd3e2e64304c71847c069f293b","subject":"only change view name from all to All when All was requested","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/b147d864468614dd3e2e64304c71847c069f293b"}]}],"author":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-17 14:08:28.000000000","tz":60},"committer":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-18 06:18:01.000000000","tz":60},"subject":"Add support for connection retry","message":"Add support for connection retry\n\nWith the new optional parameter `retries` the number of retries in case of a\nfailure can be specified.\n\nFor compability reasons the default is kept as zero.\n\nChange-Id: I9483343208d6f1be7b2b74ba993499c51852f431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/3baa5b2c3c950790899323579f4fd2c0ea85cd86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/3baa5b2c3c950790899323579f4fd2c0ea85cd86"}]},"branch":"refs/heads/master"},"6f4f86a7044709bab8d76088a9eba9de72debafd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-04-12 17:52:43.000000000","uploader":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"real_uploader":{"_account_id":34829,"name":"guillaume de mengin","display_name":"gdemengin","email":"gdemengin@gmail.com","username":"gdemengin"},"ref":"refs/changes/79/944779/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/79/944779/4","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/4"}}},"commit":{"parents":[{"commit":"ec2ce030b58c378dc2d5b0770112bc4f47b26226","subject":"fix zuul tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/ec2ce030b58c378dc2d5b0770112bc4f47b26226"}]}],"author":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-17 14:08:28.000000000","tz":60},"committer":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-04-12 17:52:43.000000000","tz":0},"subject":"Add support for connection retry","message":"Add support for connection retry\n\nWith the new optional parameter `retries` the number of retries in case of a\nfailure can be specified.\n\nFor compability reasons the default is kept as zero.\n\nChange-Id: I9483343208d6f1be7b2b74ba993499c51852f431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/6f4f86a7044709bab8d76088a9eba9de72debafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/6f4f86a7044709bab8d76088a9eba9de72debafd"}]},"branch":"refs/heads/master","description":"Rebase"},"f29d64f991b7d44140863e0f4efa17633793e949":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-04-15 18:39:34.000000000","uploader":{"_account_id":37838,"name":"Jonas Scharpf","display_name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","username":"jonas.scharpf"},"real_uploader":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"},"ref":"refs/changes/79/944779/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/jjb/python-jenkins","ref":"refs/changes/79/944779/5","commands":{"Checkout":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/jjb/python-jenkins refs/changes/79/944779/5"}}},"commit":{"parents":[{"commit":"e0b7e3380040d71532fdf3bdaf226f892b69a40d","subject":"add support for python 3.12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/e0b7e3380040d71532fdf3bdaf226f892b69a40d"}]}],"author":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-03-17 14:08:28.000000000","tz":60},"committer":{"name":"Jonas Scharpf","email":"jonas.scharpf@checkmk.com","date":"2025-04-15 18:39:34.000000000","tz":0},"subject":"Add support for connection retry","message":"Add support for connection retry\n\nWith the new optional parameter `retries` the number of retries in case of a\nfailure can be specified.\n\nFor compability reasons the default is kept as zero.\n\nChange-Id: I9483343208d6f1be7b2b74ba993499c51852f431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/f29d64f991b7d44140863e0f4efa17633793e949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/jjb/python-jenkins/commit/f29d64f991b7d44140863e0f4efa17633793e949"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34567,"name":"Vsevolod Fedorov","email":"vsevolod.fedorov@gmail.com","username":"vsevolod-fedorov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
