)]}'
{"id":"opendev%2Fpuppet-log_processor~677105","triplet_id":"opendev%2Fpuppet-log_processor~master~I5938dcef5550d4c62c8158c5f89ace75ae99aedc","project":"opendev/puppet-log_processor","branch":"master","topic":"swift-logs","hashtags":[],"change_id":"I5938dcef5550d4c62c8158c5f89ace75ae99aedc","subject":"log-gearman-worker: Remove jenkins streaming workaround","status":"MERGED","created":"2019-08-19 02:49:39.000000000","updated":"2019-08-19 15:39:56.000000000","submitted":"2019-08-19 15:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677105-1566229196676-ed945153","meta_rev_id":"300ac215bae9c3120a8c69044b2e4a486d55e098","_number":677105,"virtual_id_number":677105,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-08-19 15:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-08-19 15:18:49.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-08-19 15:13:29.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-08-19 15:18:49.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 15:13:29.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-08-19 15:18:49.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-08-19 15:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93200148dbe0d0e976b5733d4a869f480b9685dc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-19 02:49:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74cf28e94748101292c53bd4053ba1fc9fdde7e9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-19 03:07:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"320440c367eca25f417af8893d019848afe57730","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 03:29:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/af7df8dafc23411b9b0ceff18185b2a7 : SUCCESS in 7m 02s\n- legacy-puppet-syntax-4-centos-7 https://zuul.opendev.org/t/openstack/build/a716fb47a5184cff9f16b3c6d3aa02a0 : SUCCESS in 10m 30s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c8c703228a8341a2b712a23accaf8012 : SUCCESS in 12m 06s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/6654464236474d3d9946205d05768d67 : SUCCESS in 18m 39s\n- puppet-beaker-rspec-puppet-4-centos-7-infra https://zuul.opendev.org/t/openstack/build/ec63a972be2443e8a4bef17772c88685 : SUCCESS in 14m 23s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/dee0796a56a04ddca60ecf782fe7cbc9 : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":2},{"id":"22a90b504512b8e6d3634eb7ab9fc771e5154060","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-19 07:17:29.000000000","message":"Topic changed from deflate to swift-logs","accounts_in_message":[],"_revision_number":2},{"id":"f468398f5905e455b972b2d1f8214cd78294272c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-19 15:13:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"085851e4456168eac8c04605b9f95774442f8e78","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-19 15:18:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"69d9468f5df2127b3b5237628a5c72cf916a5a9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:19:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"116e7bb2ac43e1e871f60c93d06959a04b5dc80a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:39:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/3231306b09b8499999b9151b70de5a77 : SUCCESS in 9m 35s\n- legacy-puppet-syntax-4-centos-7 https://zuul.opendev.org/t/openstack/build/89ef519a7d7c45df8c749b8f6f62c515 : SUCCESS in 9m 12s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6121153a526741fa867decd7c8af4736 : SUCCESS in 13m 20s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/28a8df39af224dbc8b1cd09ff59898f2 : SUCCESS in 14m 17s\n- puppet-beaker-rspec-puppet-4-centos-7-infra https://zuul.opendev.org/t/openstack/build/1608f0fbf25b4b929b862d99c39afd64 : SUCCESS in 11m 43s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/568bfdac230546679a0fb2e0024a66dd : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":2},{"id":"300ac215bae9c3120a8c69044b2e4a486d55e098","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:39:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b30a3a6c0355f1b9f02c1e21d64dbb060ee9622","revisions":{"60b033abe0d336fe0eeeecea1d331f21f38e78cb":{"kind":"REWORK","_number":1,"created":"2019-08-19 02:49:39.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/05/677105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-log_processor","ref":"refs/changes/05/677105/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/1"}}},"commit":{"parents":[{"commit":"a0d85c77ae964bd866a8028804390a5fc0b6a75f","subject":"log-gearman-worker: handle deflate encoded values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/a0d85c77ae964bd866a8028804390a5fc0b6a75f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-19 02:25:52.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-19 02:49:15.000000000","tz":600},"subject":"log-gearman-worker: Remove jenkins streaming workaround","message":"log-gearman-worker: Remove jenkins streaming workaround\n\nWe don\u0027t need to worry about the file changing under us any more; this\nwas all pre-zuul, let alone pre-using-swift for logs.  Remove this\nworkaround.\n\nChange-Id: I5938dcef5550d4c62c8158c5f89ace75ae99aedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/60b033abe0d336fe0eeeecea1d331f21f38e78cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/60b033abe0d336fe0eeeecea1d331f21f38e78cb"}]},"branch":"refs/heads/master"},"5b30a3a6c0355f1b9f02c1e21d64dbb060ee9622":{"kind":"NO_CHANGE","_number":2,"created":"2019-08-19 03:07:19.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/05/677105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-log_processor","ref":"refs/changes/05/677105/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-log_processor refs/changes/05/677105/2"}}},"commit":{"parents":[{"commit":"bca04e3155c7b7c5318618d79cacb4243461ddc2","subject":"log-gearman-worker: handle deflate encoded values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/bca04e3155c7b7c5318618d79cacb4243461ddc2"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-19 02:25:52.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-19 03:06:58.000000000","tz":600},"subject":"log-gearman-worker: Remove jenkins streaming workaround","message":"log-gearman-worker: Remove jenkins streaming workaround\n\nWe don\u0027t need to worry about the file changing under us any more; this\nwas all pre-zuul, let alone pre-using-swift for logs.  Remove this\nworkaround.\n\nChange-Id: I5938dcef5550d4c62c8158c5f89ace75ae99aedc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/5b30a3a6c0355f1b9f02c1e21d64dbb060ee9622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/5b30a3a6c0355f1b9f02c1e21d64dbb060ee9622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
