)]}'
{"id":"opendev%2Fpuppet-log_processor~677260","triplet_id":"opendev%2Fpuppet-log_processor~master~I7ab56ac397133f00539d9d3374fa400363ef12d6","project":"opendev/puppet-log_processor","branch":"master","topic":"swift-logs","hashtags":[],"change_id":"I7ab56ac397133f00539d9d3374fa400363ef12d6","subject":"Fix systemd severity filter input data","status":"MERGED","created":"2019-08-19 17:27:51.000000000","updated":"2019-08-19 19:41:36.000000000","submitted":"2019-08-19 19:41:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677260-1566243696720-de1e3c07","meta_rev_id":"501123fc671ddbecf66a35bee92c7522d61ee704","_number":677260,"virtual_id_number":677260,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-08-19 19:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-08-19 19:09:47.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-19 17:29:53.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-08-19 19:09:47.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 17:29:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-19 19:09:47.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-08-19 19:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49632279773a3a15b20e15b6eb87d940c56db01c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-19 17:27:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9426ed573f088de45348eaae03873036acaf2319","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-19 17:29:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d1d328bc792b8349121405a1858f33d4dbd8f6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:49:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/c655156c396a440dba0a12eb69e45bb0 : SUCCESS in 7m 25s\n- legacy-puppet-syntax-4-centos-7 https://zuul.opendev.org/t/openstack/build/2c43ef403faa4d70865f706bbf0a5a9e : SUCCESS in 8m 55s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/4fdb3d3b875441fdbd8d6e50dc8c0d14 : SUCCESS in 14m 03s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/1893113404894dcf9c9f0b611f56bd1b : SUCCESS in 15m 26s\n- puppet-beaker-rspec-puppet-4-centos-7-infra https://zuul.opendev.org/t/openstack/build/d4f223171fbd4186adb2d8f8bc1868bd : SUCCESS in 15m 05s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/346dc3578cb8435d8cfb89fc0c3c838f : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ceaaad889fb3399c8dd934a7143f1eb1f663df3a","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-19 19:09:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe09fddd503c2ae5b97de36b8a159d084e112c20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 19:09:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29fd7e4beada955eec69729708d4dee08f355a92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 19:41:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/43543f27dedc424f9890574bffc1d173 : SUCCESS in 6m 05s\n- legacy-puppet-syntax-4-centos-7 https://zuul.opendev.org/t/openstack/build/f496c1f2e7cd4d08b7efb734eec5ddad : SUCCESS in 9m 59s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fc8b136e0e6145bcaf4062acfbd03e1f : SUCCESS in 16m 52s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/bf129a2d3ccb44dab0efc9e73fb8c4fa : SUCCESS in 25m 31s\n- puppet-beaker-rspec-puppet-4-centos-7-infra https://zuul.opendev.org/t/openstack/build/e7b0362bd8fd4b42a61e5daeca304a7c : SUCCESS in 14m 16s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/d1a57c8f24ea4824a110c2cc492fc18d : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"25835c6ee760569e112b726721a65a9f94ed2a54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 19:41:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9063a7e7ed6e4313d792c443378552f5b2a2726","revisions":{"b9063a7e7ed6e4313d792c443378552f5b2a2726":{"kind":"REWORK","_number":1,"created":"2019-08-19 17:27:51.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/60/677260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-log_processor","ref":"refs/changes/60/677260/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/60/677260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/60/677260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-log_processor refs/changes/60/677260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-log_processor refs/changes/60/677260/1"}}},"commit":{"parents":[{"commit":"3119c0cddd86d74cb554d0338e6f0125779ea819","subject":"log-gearman-worker: remove obsolete GET debug filter, add local filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/3119c0cddd86d74cb554d0338e6f0125779ea819"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-08-19 17:26:20.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-08-19 17:27:45.000000000","tz":-420},"subject":"Fix systemd severity filter input data","message":"Fix systemd severity filter input data\n\nThe severity filters are passed the entire json event and not just a\nstring. Update the systemd filter to access the message string out of\nthe event json dict.\n\nPrior to this we get a type error:\n\n  2019-08-19 17:18:48,055 Exception handling log event.\n  Traceback (most recent call last):\n    File \"/usr/local/bin/log-gearman-worker.py\", line 255, in\n  _handle_event\n      keep_line \u003d f.process(out_event)\n    File \"/usr/local/bin/log-gearman-worker.py\", line 183, in process\n      m \u003d self.SYSTEMDRE.match(msg)\n  TypeError: expected string or buffer\n\nChange-Id: I7ab56ac397133f00539d9d3374fa400363ef12d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/b9063a7e7ed6e4313d792c443378552f5b2a2726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-log_processor/commit/b9063a7e7ed6e4313d792c443378552f5b2a2726"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
