)]}'
{"id":"opendev%2Fpuppet-openstackci~658148","triplet_id":"opendev%2Fpuppet-openstackci~master~I2ae4fe3f43e5e0e606a58f58a4f6e2bafd3a3930","project":"opendev/puppet-openstackci","branch":"master","hashtags":[],"change_id":"I2ae4fe3f43e5e0e606a58f58a4f6e2bafd3a3930","subject":"Pin keyring to 18.0.1","status":"MERGED","created":"2019-05-09 16:10:13.000000000","updated":"2019-05-09 20:14:13.000000000","submitted":"2019-05-09 20:14:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658148-1557432853620-e62a4e3b","meta_rev_id":"6d06e4e77d8875ba3e04f9e91ebbafee5caa3673","_number":658148,"virtual_id_number":658148,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":2,"date":"2019-05-09 20:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-05-09 16:19:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-09 19:26:09.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":2,"date":"2019-05-09 19:30:24.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":1,"date":"2019-05-09 19:30:24.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 16:19:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-09 19:26:09.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-05-09 19:30:24.000000000","updated_by":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"reviewer":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"state":"REVIEWER"},{"updated":"2019-05-09 20:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac30428d722ffa5dc33a8474d8f1a1a0fc593c55","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-09 16:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc3817090b240f69f5c5c497ae7a964a0ca641d7","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-05-09 16:15:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"305fd8c777f9824eefa4bf8aaaa73099c41b5e2a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-09 16:19:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"20fa592316d16427dac7897c9c31c9379f103c1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 16:45:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/48/658148/2/check/legacy-puppet-lint/0ea108a/ : SUCCESS in 3m 59s\n- legacy-puppet-syntax-3 http://logs.openstack.org/48/658148/2/check/legacy-puppet-syntax-3/724d8d0/ : SUCCESS in 5m 24s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/48/658148/2/check/legacy-puppet-syntax-4-centos-7/17ce6ce/ : SUCCESS in 6m 56s\n- infra-puppet-apply-3-ubuntu-trusty http://logs.openstack.org/48/658148/2/check/infra-puppet-apply-3-ubuntu-trusty/179e241/ : SUCCESS in 6m 17s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/48/658148/2/check/infra-puppet-apply-3-ubuntu-xenial/2eb2c9a/ : SUCCESS in 9m 55s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/48/658148/2/check/infra-puppet-apply-4-ubuntu-xenial/845908c/ : SUCCESS in 11m 39s\n- openstackci-beaker-ubuntu-trusty http://logs.openstack.org/48/658148/2/check/openstackci-beaker-ubuntu-trusty/7f4ee11/ : SUCCESS in 14m 17s\n- openstackci-beaker http://logs.openstack.org/48/658148/2/check/openstackci-beaker/65b80e7/ : SUCCESS in 11m 56s\n- openstackci-beaker-puppet-4 http://logs.openstack.org/48/658148/2/check/openstackci-beaker-puppet-4/db8f782/ : SUCCESS in 12m 57s (non-voting)\n- tox-docs http://logs.openstack.org/48/658148/2/check/tox-docs/9ee41a1/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c443b444fb22cef2ae025483adcd77d35b29937a","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2019-05-09 19:26:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18926ec3518bed61ed649dc1eda17d6014eff1e5","author":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"date":"2019-05-09 19:30:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6158486b56ed4351670ea9eaca8044ad2e0acc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 19:30:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7799b733e9e7430757fb2e7bac6e84ba0c65087c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:14:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/48/658148/2/gate/legacy-puppet-lint/5d60122/ : SUCCESS in 5m 48s\n- legacy-puppet-syntax-3 http://logs.openstack.org/48/658148/2/gate/legacy-puppet-syntax-3/3a87f70/ : SUCCESS in 7m 10s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/48/658148/2/gate/legacy-puppet-syntax-4-centos-7/01735f4/ : SUCCESS in 7m 18s\n- infra-puppet-apply-3-ubuntu-trusty http://logs.openstack.org/48/658148/2/gate/infra-puppet-apply-3-ubuntu-trusty/8708de0/ : SUCCESS in 6m 46s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/48/658148/2/gate/infra-puppet-apply-3-ubuntu-xenial/f0a1df6/ : SUCCESS in 10m 14s\n- openstackci-beaker-ubuntu-trusty http://logs.openstack.org/48/658148/2/gate/openstackci-beaker-ubuntu-trusty/8069efd/ : SUCCESS in 16m 51s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/48/658148/2/gate/infra-puppet-apply-4-ubuntu-xenial/f4b2c68/ : SUCCESS in 10m 48s\n- openstackci-beaker http://logs.openstack.org/48/658148/2/gate/openstackci-beaker/cbf3fba/ : SUCCESS in 12m 29s\n- tox-docs http://logs.openstack.org/48/658148/2/gate/tox-docs/d90cecd/html/ : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2},{"id":"6d06e4e77d8875ba3e04f9e91ebbafee5caa3673","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:14:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b42b3d4e5bd2ddb3cc5d83eaf1332471f530c5f4","revisions":{"471570035b4c6eff3e80c6c97bd5b51ba7279c0c":{"kind":"REWORK","_number":1,"created":"2019-05-09 16:10:13.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/48/658148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/48/658148/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/1"}}},"commit":{"parents":[{"commit":"05d6bd00bf7b5f8eea2d9cc47b33ef50e3505665","subject":"Fix opendev paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/05d6bd00bf7b5f8eea2d9cc47b33ef50e3505665"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-09 16:09:39.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-09 16:09:39.000000000","tz":0},"subject":"Pin keyring to 18.0.0","message":"Pin keyring to 18.0.0\n\nkeyring 19.0.0 only supports python3. While we should obviously\nget things updated, we\u0027re going to need to unwind some things\nto do that.\n\nChange-Id: I2ae4fe3f43e5e0e606a58f58a4f6e2bafd3a3930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/471570035b4c6eff3e80c6c97bd5b51ba7279c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/471570035b4c6eff3e80c6c97bd5b51ba7279c0c"}]},"branch":"refs/heads/master"},"b42b3d4e5bd2ddb3cc5d83eaf1332471f530c5f4":{"kind":"REWORK","_number":2,"created":"2019-05-09 16:15:49.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/48/658148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/48/658148/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/48/658148/2"}}},"commit":{"parents":[{"commit":"05d6bd00bf7b5f8eea2d9cc47b33ef50e3505665","subject":"Fix opendev paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/05d6bd00bf7b5f8eea2d9cc47b33ef50e3505665"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-09 16:09:39.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-05-09 16:15:41.000000000","tz":0},"subject":"Pin keyring to 18.0.1","message":"Pin keyring to 18.0.1\n\nkeyring 19.0.0 only supports python3. While we should obviously\nget things updated, we\u0027re going to need to unwind some things\nto do that.\n\nChange-Id: I2ae4fe3f43e5e0e606a58f58a4f6e2bafd3a3930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/b42b3d4e5bd2ddb3cc5d83eaf1332471f530c5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/b42b3d4e5bd2ddb3cc5d83eaf1332471f530c5f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
