)]}'
{"id":"opendev%2Fpuppet-openstackci~661341","triplet_id":"opendev%2Fpuppet-openstackci~master~I1092b3a2f4b781465c2148336076677d3c07211e","project":"opendev/puppet-openstackci","branch":"master","topic":"no-idempotency-checks","hashtags":[],"change_id":"I1092b3a2f4b781465c2148336076677d3c07211e","subject":"Remove idempotency spec checks","status":"MERGED","created":"2019-05-24 15:41:31.000000000","updated":"2019-05-24 17:28:31.000000000","submitted":"2019-05-24 17:28:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661341-1558718911385-77ae9cd1","meta_rev_id":"4457e7eb1aa1fd952f24c27b0917fab712b53f04","_number":661341,"virtual_id_number":661341,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2019-05-24 17:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-05-24 17:09:20.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-24 17:10:05.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-24 17:10:05.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 17:09:20.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-05-24 17:10:05.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-05-24 17:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81aa950a5c1063133c4955340e0f2a9ebbf05d72","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-05-24 15:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b0ef1f013989b6e5285114bc17b9290f15c65b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:17:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/41/661341/1/check/legacy-puppet-lint/0961008/ : SUCCESS in 4m 24s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/41/661341/1/check/legacy-puppet-syntax-4-centos-7/8a16c74/ : SUCCESS in 21m 16s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/41/661341/1/check/infra-puppet-apply-3-ubuntu-xenial/f7403e6/ : SUCCESS in 10m 55s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/41/661341/1/check/infra-puppet-apply-4-ubuntu-xenial/92822c7/ : SUCCESS in 9m 09s\n- openstackci-beaker-puppet-4 http://logs.openstack.org/41/661341/1/check/openstackci-beaker-puppet-4/cecf999/ : SUCCESS in 14m 13s\n- tox-docs http://logs.openstack.org/41/661341/1/check/tox-docs/30f6ab3/html/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d707b91da66a309f42d55b30ebef6774688bf650","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-05-24 17:09:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f425e6ededeb4438cd7207a4c889999b569ba0d6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-24 17:10:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bae81612fd96c639489d5090ee61ab43e67fe52c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:10:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dd1eae39a2eb917cd414001ebc8202446153d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:28:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92399f97fda4702f70e3ade2b6e45e74fa0868ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:28:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/41/661341/1/gate/legacy-puppet-lint/483860d/ : SUCCESS in 4m 18s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/41/661341/1/gate/legacy-puppet-syntax-4-centos-7/09c85e8/ : SUCCESS in 6m 45s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/41/661341/1/gate/infra-puppet-apply-3-ubuntu-xenial/3e5941c/ : SUCCESS in 7m 53s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/41/661341/1/gate/infra-puppet-apply-4-ubuntu-xenial/12cc068/ : SUCCESS in 9m 16s\n- openstackci-beaker-puppet-4 http://logs.openstack.org/41/661341/1/gate/openstackci-beaker-puppet-4/c4ab31c/ : SUCCESS in 11m 14s\n- tox-docs http://logs.openstack.org/41/661341/1/gate/tox-docs/7e4f620/html/ : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e266e8498279aaf257bb2789159eda9192e9c997","revisions":{"e266e8498279aaf257bb2789159eda9192e9c997":{"kind":"REWORK","_number":1,"created":"2019-05-24 15:41:31.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/41/661341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/41/661341/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/41/661341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/41/661341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/41/661341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/41/661341/1"}}},"commit":{"parents":[{"commit":"fd385b1210e6a862e813c7e292adcf3bc3a14d9f","subject":"The logs for a job my contain dotfiles, index them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/fd385b1210e6a862e813c7e292adcf3bc3a14d9f"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-05-24 15:40:38.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-05-24 15:40:38.000000000","tz":-420},"subject":"Remove idempotency spec checks","message":"Remove idempotency spec checks\n\nThe idempotency checks race external changes like updates to\nproject-config or nodepool. This often causes this job to fail when we\nare busy updating other parts of our sytems.\n\nIdempotency is nice but not necessary. Lets remove these checks to make\ntesting more reliable.\n\nChange-Id: I1092b3a2f4b781465c2148336076677d3c07211e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/e266e8498279aaf257bb2789159eda9192e9c997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/e266e8498279aaf257bb2789159eda9192e9c997"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
