)]}'
{"id":"opendev%2Fpuppet-openstackci~674764","triplet_id":"opendev%2Fpuppet-openstackci~master~Id34c74d848638182819a9c81d051b83820244463","project":"opendev/puppet-openstackci","branch":"master","topic":"symlinks","hashtags":[],"change_id":"Id34c74d848638182819a9c81d051b83820244463","subject":"Log archive maintenance; also clear out symlinks","status":"MERGED","created":"2019-08-06 09:16:09.000000000","updated":"2019-08-06 13:56:00.000000000","submitted":"2019-08-06 13:56:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674764-1565099760211-a2196f10","meta_rev_id":"b1a2b80a432ba5578f7b3b2fdc623a4baa533ac8","_number":674764,"virtual_id_number":674764,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-06 13:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-08-06 13:32:34.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-08-06 12:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-06 13:36:01.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-06 13:36:01.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 12:59:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-06 13:32:34.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-08-06 13:36:01.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-06 13:56:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22186cc818dc1afe3c50b78f425f043f031338b3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-06 09:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b155f8af91e42290c0d857d76cddc5a7fa6f186e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 09:41:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://logs.opendev.org/64/674764/1/check/legacy-puppet-lint/9928237/ : SUCCESS in 6m 07s\n- legacy-puppet-syntax-4-centos-7 https://logs.opendev.org/64/674764/1/check/legacy-puppet-syntax-4-centos-7/aa12e91/ : SUCCESS in 12m 20s\n- infra-puppet-apply-4-ubuntu-xenial https://logs.opendev.org/64/674764/1/check/infra-puppet-apply-4-ubuntu-xenial/180de04/ : SUCCESS in 17m 49s\n- openstackci-beaker-puppet-4 https://logs.opendev.org/64/674764/1/check/openstackci-beaker-puppet-4/c90e87c/ : SUCCESS in 20m 05s\n- tox-docs https://logs.opendev.org/64/674764/1/check/tox-docs/2fcaa04/html/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8616edfe791b310597900f7124bada18291ad6ad","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-06 12:59:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"903ae456d4047da3d6d63800c0f791c9496c6e3f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-08-06 13:32:34.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM - but I definitely want fungi to look at it before we land it.","accounts_in_message":[],"_revision_number":1},{"id":"c189f6128a807fa0f9a339de33246186ddfcfb6d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-06 13:36:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3457aae4585ed2559b2a7accbc43803d94fc8bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:36:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"133189bc995562de123e395ce6dad6a524e6662a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:56:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf480877549ea6fdddf484faf9909a772336f2aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:56:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://logs.opendev.org/64/674764/1/gate/legacy-puppet-lint/7dd554b/ : SUCCESS in 5m 54s\n- legacy-puppet-syntax-4-centos-7 https://logs.opendev.org/64/674764/1/gate/legacy-puppet-syntax-4-centos-7/2d37477/ : SUCCESS in 7m 31s\n- infra-puppet-apply-4-ubuntu-xenial https://logs.opendev.org/64/674764/1/gate/infra-puppet-apply-4-ubuntu-xenial/ba1c5f6/ : SUCCESS in 16m 28s\n- openstackci-beaker-puppet-4 https://logs.opendev.org/64/674764/1/gate/openstackci-beaker-puppet-4/622f3ca/ : SUCCESS in 12m 45s\n- tox-docs https://logs.opendev.org/64/674764/1/gate/tox-docs/a660e58/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"329f330b3efbb345a599a5de4ea2fd7408f24262","revisions":{"329f330b3efbb345a599a5de4ea2fd7408f24262":{"kind":"REWORK","_number":1,"created":"2019-08-06 09:16:09.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/64/674764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/64/674764/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/64/674764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/64/674764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/64/674764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/64/674764/1"}}},"commit":{"parents":[{"commit":"f4c637c73881350d57002e4536cf5e0e44bbe739","subject":"Pin ara at 0.16.5 for logserver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/f4c637c73881350d57002e4536cf5e0e44bbe739"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-06 09:13:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2019-08-06 09:13:44.000000000","tz":600},"subject":"Log archive maintenance; also clear out symlinks","message":"Log archive maintenance; also clear out symlinks\n\nCurrently if a log directory has a symlink it is skipped in the find;\nthus the directory does not become empty and it is never removed.\nAlso match symlinks.\n\nChange-Id: Id34c74d848638182819a9c81d051b83820244463\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/329f330b3efbb345a599a5de4ea2fd7408f24262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/329f330b3efbb345a599a5de4ea2fd7408f24262"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
