)]}'
{"id":"opendev%2Fpuppet-openstackci~674920","triplet_id":"opendev%2Fpuppet-openstackci~master~I369565a780eb749eab95e4bf8e7d8d7fd5bf0630","project":"opendev/puppet-openstackci","branch":"master","topic":"robots","hashtags":[],"change_id":"I369565a780eb749eab95e4bf8e7d8d7fd5bf0630","subject":"Handle robots.txt on logs.o.o","status":"MERGED","created":"2019-08-06 19:35:57.000000000","updated":"2019-08-07 15:19:25.000000000","submitted":"2019-08-07 15:19:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674920-1565191165220-e657b839","meta_rev_id":"3338c9d3573ddf51b64f0534f6b52dc4cb02244a","_number":674920,"virtual_id_number":674920,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-08-07 14:05:08.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-07 15:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2019-08-07 14:07:17.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2019-08-07 14:11:48.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2019-08-07 14:11:48.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 05:32:30.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-08-07 14:07:17.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-08-07 14:11:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-07 15:19:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"716a0958d45bb68271777b30de7c5451f7b185c6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-06 19:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb057bb132447fd1669d648117b30166c263e331","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:01:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://logs.opendev.org/20/674920/1/check/legacy-puppet-lint/28ca6f1/ : SUCCESS in 5m 08s\n- legacy-puppet-syntax-4-centos-7 https://logs.opendev.org/20/674920/1/check/legacy-puppet-syntax-4-centos-7/6eb0c89/ : SUCCESS in 10m 29s\n- infra-puppet-apply-4-ubuntu-xenial https://logs.opendev.org/20/674920/1/check/infra-puppet-apply-4-ubuntu-xenial/ce574f8/ : SUCCESS in 13m 17s\n- openstackci-beaker-puppet-4 https://logs.opendev.org/20/674920/1/check/openstackci-beaker-puppet-4/bb62008/ : SUCCESS in 14m 36s\n- tox-docs https://logs.opendev.org/20/674920/1/check/tox-docs/2df4988/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2797edc047de15f9765048dc2dd49da2090c4d89","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-06 20:03:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24295901048a97f9e6a1238cda0b5ee79007fb67","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 04:21:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d946e2affbf0bf87ea54b4889a3e4804fd76c22e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 04:34:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"715e930436ce2afb66d732c0f5e50129f57b08e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 05:01:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://logs.opendev.org/20/674920/3/check/legacy-puppet-lint/9adb1f8/ : SUCCESS in 5m 31s\n- legacy-puppet-syntax-4-centos-7 https://logs.opendev.org/20/674920/3/check/legacy-puppet-syntax-4-centos-7/aaf1dd9/ : SUCCESS in 8m 46s\n- infra-puppet-apply-4-ubuntu-xenial https://logs.opendev.org/20/674920/3/check/infra-puppet-apply-4-ubuntu-xenial/d00fb48/ : SUCCESS in 15m 14s\n- openstackci-beaker-puppet-4 https://logs.opendev.org/20/674920/3/check/openstackci-beaker-puppet-4/e22ed53/ : SUCCESS in 16m 57s\n- tox-docs https://logs.opendev.org/20/674920/3/check/tox-docs/ed26b04/html/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"a9d05defc4409dfb42d1c0a9ab19a909495963c0","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-07 05:32:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6e7de7fff9d6df6982efcd4e0f104aa593692b5a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-07 14:03:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba00ced5846199da42a46a36a098d0c1c2c3697d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 14:04:59.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"46607f1d9907d484c44711e1baebc41ce7a4426d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-07 14:05:08.000000000","message":"Patch Set 4:\n\nThanks for catching that!","accounts_in_message":[],"_revision_number":4},{"id":"69d82ab9361bddcdbae7b902cc743d1e8fd1a811","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-08-07 14:07:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1fb1626e7a953dec69a03520692d9c252028845e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-07 14:11:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"148e54c0c4c4024a6768e649b6b1bc1f51844dc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:49:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://logs.opendev.org/20/674920/4/check/legacy-puppet-lint/82425cc/ : SUCCESS in 4m 21s\n- legacy-puppet-syntax-4-centos-7 https://logs.opendev.org/20/674920/4/check/legacy-puppet-syntax-4-centos-7/bec9494/ : SUCCESS in 9m 30s\n- infra-puppet-apply-4-ubuntu-xenial https://logs.opendev.org/20/674920/4/check/infra-puppet-apply-4-ubuntu-xenial/cab7346/ : SUCCESS in 8m 45s\n- openstackci-beaker-puppet-4 https://logs.opendev.org/20/674920/4/check/openstackci-beaker-puppet-4/712fcf6/ : SUCCESS in 14m 18s\n- tox-docs https://logs.opendev.org/20/674920/4/check/tox-docs/c4f3f96/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":4},{"id":"3e80c6168d903f6d74e7ee48c219cb340007d47e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:49:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7479c8f6c20010e64f84aed3df4a5a7079bbc37d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 15:19:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://logs.opendev.org/20/674920/4/gate/legacy-puppet-lint/264b98f/ : SUCCESS in 6m 06s\n- legacy-puppet-syntax-4-centos-7 https://logs.opendev.org/20/674920/4/gate/legacy-puppet-syntax-4-centos-7/8bc7c0d/ : SUCCESS in 11m 07s\n- infra-puppet-apply-4-ubuntu-xenial https://logs.opendev.org/20/674920/4/gate/infra-puppet-apply-4-ubuntu-xenial/39e644a/ : SUCCESS in 18m 15s\n- openstackci-beaker-puppet-4 https://logs.opendev.org/20/674920/4/gate/openstackci-beaker-puppet-4/88499cb/ : SUCCESS in 21m 41s\n- tox-docs https://logs.opendev.org/20/674920/4/gate/tox-docs/651fdde/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"dbd52779d3cb569f42d68b37de2d55c2981a31f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 15:19:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f494aa87626d09b809e1727fbc96fe3a003167ca","revisions":{"96a897774f541906e987ccfd84c7fd000f892543":{"kind":"REWORK","_number":1,"created":"2019-08-06 19:35:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/674920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/20/674920/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/1"}}},"commit":{"parents":[{"commit":"329f330b3efbb345a599a5de4ea2fd7408f24262","subject":"Log archive maintenance; also clear out symlinks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/329f330b3efbb345a599a5de4ea2fd7408f24262"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-06 19:32:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-06 19:35:47.000000000","tz":120},"subject":"Handle robots.txt on logs.o.o","message":"Handle robots.txt on logs.o.o\n\nWe don\u0027t want to delete the top-level robots.txt file - but everything\nthat is part of the logs hierarchy.\n\nAdd an extra find run to remove robots.txt. Note that we don\u0027t compress\nsince it\u0027s very small (200 bytes).\n\nChange-Id: I369565a780eb749eab95e4bf8e7d8d7fd5bf0630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/96a897774f541906e987ccfd84c7fd000f892543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/96a897774f541906e987ccfd84c7fd000f892543"}]},"branch":"refs/heads/master"},"8f2be09fc2f801c131dca6230eddb8a9dee5c792":{"kind":"REWORK","_number":2,"created":"2019-08-07 04:21:33.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/674920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/20/674920/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/2"}}},"commit":{"parents":[{"commit":"329f330b3efbb345a599a5de4ea2fd7408f24262","subject":"Log archive maintenance; also clear out symlinks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/329f330b3efbb345a599a5de4ea2fd7408f24262"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-06 19:32:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 04:20:50.000000000","tz":120},"subject":"Handle robots.txt on logs.o.o","message":"Handle robots.txt on logs.o.o\n\nWe don\u0027t want to delete the top-level robots.txt file - but everything\nthat is part of the logs hierarchy.\n\nChange-Id: I369565a780eb749eab95e4bf8e7d8d7fd5bf0630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/8f2be09fc2f801c131dca6230eddb8a9dee5c792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/8f2be09fc2f801c131dca6230eddb8a9dee5c792"}]},"branch":"refs/heads/master"},"80cf010ede4a2ed1d5329b0728492e7db025859d":{"kind":"REWORK","_number":3,"created":"2019-08-07 04:34:13.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/674920/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/20/674920/3","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/3"}}},"commit":{"parents":[{"commit":"329f330b3efbb345a599a5de4ea2fd7408f24262","subject":"Log archive maintenance; also clear out symlinks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/329f330b3efbb345a599a5de4ea2fd7408f24262"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-06 19:32:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-07 04:34:05.000000000","tz":120},"subject":"Handle robots.txt on logs.o.o","message":"Handle robots.txt on logs.o.o\n\nWe don\u0027t want to delete the top-level robots.txt file - but everything\nthat is part of the logs hierarchy.\n\nChange-Id: I369565a780eb749eab95e4bf8e7d8d7fd5bf0630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/80cf010ede4a2ed1d5329b0728492e7db025859d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/80cf010ede4a2ed1d5329b0728492e7db025859d"}]},"branch":"refs/heads/master"},"f494aa87626d09b809e1727fbc96fe3a003167ca":{"kind":"REWORK","_number":4,"created":"2019-08-07 14:04:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/20/674920/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackci","ref":"refs/changes/20/674920/4","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackci refs/changes/20/674920/4"}}},"commit":{"parents":[{"commit":"329f330b3efbb345a599a5de4ea2fd7408f24262","subject":"Log archive maintenance; also clear out symlinks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/329f330b3efbb345a599a5de4ea2fd7408f24262"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-06 19:32:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-07 14:04:57.000000000","tz":0},"subject":"Handle robots.txt on logs.o.o","message":"Handle robots.txt on logs.o.o\n\nWe don\u0027t want to delete the top-level robots.txt file - but everything\nthat is part of the logs hierarchy.\n\nChange-Id: I369565a780eb749eab95e4bf8e7d8d7fd5bf0630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/f494aa87626d09b809e1727fbc96fe3a003167ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackci/commit/f494aa87626d09b809e1727fbc96fe3a003167ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
