)]}'
{"id":"opendev%2Fpuppet-openstackid~656411","triplet_id":"opendev%2Fpuppet-openstackid~master~I487a973826fb40675e5e9999be89d69481a7312e","project":"opendev/puppet-openstackid","branch":"master","topic":"hotfix/rewrite_rules_for_registration","hashtags":[],"change_id":"I487a973826fb40675e5e9999be89d69481a7312e","subject":"Added rewriting rules","status":"MERGED","created":"2019-04-29 21:58:22.000000000","updated":"2019-04-29 23:20:17.000000000","submitted":"2019-04-29 23:20:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656411-1556580017084-332a2ac9","meta_rev_id":"01c4543c3e0657d69828f945199c9c73e21a4ead","_number":656411,"virtual_id_number":656411,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-04-29 23:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-04-29 22:03:18.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-29 22:50:26.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-29 22:50:26.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 22:03:18.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-29 22:50:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-04-29 23:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e9798c689e1f817c1ccb6a4fdad43b12b6c9728","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2019-04-29 21:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48e8ad09d2543f806430c59ea21e94b61001bb1c","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2019-04-29 21:59:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a862b5dbb28ac2a4670c538b47d9e53dbce4be07","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-29 22:00:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"66e0b674f04ab2c2802de1afb110f233fbafeb8d","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2019-04-29 22:02:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76079cc37224792f2d72e194858a76268be5b72f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-29 22:03:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a828620f589ccfdf90c7ab4e65660c0c4feaedd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 22:22:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/11/656411/3/check/legacy-puppet-lint/63ef515/ : SUCCESS in 5m 46s\n- legacy-puppet-syntax-3 http://logs.openstack.org/11/656411/3/check/legacy-puppet-syntax-3/9b9efdf/ : SUCCESS in 4m 33s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/11/656411/3/check/legacy-puppet-syntax-4-centos-7/1d907b5/ : SUCCESS in 8m 03s\n- infra-puppet-apply-3-ubuntu-trusty http://logs.openstack.org/11/656411/3/check/infra-puppet-apply-3-ubuntu-trusty/708c882/ : SUCCESS in 7m 59s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/11/656411/3/check/infra-puppet-apply-3-ubuntu-xenial/2cc8774/ : SUCCESS in 8m 30s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/11/656411/3/check/infra-puppet-apply-4-ubuntu-xenial/f87c16f/ : SUCCESS in 9m 27s\n- openstackci-beaker-ubuntu-trusty http://logs.openstack.org/11/656411/3/check/openstackci-beaker-ubuntu-trusty/697dcdb/ : SUCCESS in 11m 58s\n- openstackci-beaker http://logs.openstack.org/11/656411/3/check/openstackci-beaker/3eeeb04/ : SUCCESS in 13m 59s\n- openstackci-beaker-puppet-4 http://logs.openstack.org/11/656411/3/check/openstackci-beaker-puppet-4/eafcf38/ : SUCCESS in 12m 26s (non-voting)\n- puppet-beaker-rspec-centos-7-infra http://logs.openstack.org/11/656411/3/check/puppet-beaker-rspec-centos-7-infra/3c54112/ : SUCCESS in 11m 42s\n- puppet-beaker-rspec-puppet-4-centos-7-infra http://logs.openstack.org/11/656411/3/check/puppet-beaker-rspec-puppet-4-centos-7-infra/ad670d4/ : SUCCESS in 8m 43s (non-voting)\n- puppet-beaker-rspec-infra http://logs.openstack.org/11/656411/3/check/puppet-beaker-rspec-infra/38b55b4/ : SUCCESS in 4m 42s\n- puppet-beaker-rspec-puppet-4-infra http://logs.openstack.org/11/656411/3/check/puppet-beaker-rspec-puppet-4-infra/1401d92/ : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"82e6cd7a49a31518973127f816b1f1fc2ecc6163","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-04-29 22:50:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"383b0efe2f79893b975295ae5275f59f7757995c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 22:50:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1e08a3ac3f122148d2684832cfae41292c45671f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 23:20:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/11/656411/3/gate/legacy-puppet-lint/4d3a0cb/ : SUCCESS in 5m 50s\n- legacy-puppet-syntax-3 http://logs.openstack.org/11/656411/3/gate/legacy-puppet-syntax-3/b304ed2/ : SUCCESS in 9m 16s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/11/656411/3/gate/legacy-puppet-syntax-4-centos-7/9f53f17/ : SUCCESS in 11m 46s\n- infra-puppet-apply-3-ubuntu-trusty http://logs.openstack.org/11/656411/3/gate/infra-puppet-apply-3-ubuntu-trusty/aded003/ : SUCCESS in 11m 43s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/11/656411/3/gate/infra-puppet-apply-3-ubuntu-xenial/73c7445/ : SUCCESS in 15m 36s\n- openstackci-beaker-ubuntu-trusty http://logs.openstack.org/11/656411/3/gate/openstackci-beaker-ubuntu-trusty/7b017ae/ : SUCCESS in 15m 31s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/11/656411/3/gate/infra-puppet-apply-4-ubuntu-xenial/5262911/ : SUCCESS in 15m 05s\n- openstackci-beaker http://logs.openstack.org/11/656411/3/gate/openstackci-beaker/e5fcc62/ : SUCCESS in 18m 52s\n- puppet-beaker-rspec-centos-7-infra http://logs.openstack.org/11/656411/3/gate/puppet-beaker-rspec-centos-7-infra/f573e58/ : SUCCESS in 8m 29s\n- puppet-beaker-rspec-infra http://logs.openstack.org/11/656411/3/gate/puppet-beaker-rspec-infra/28f7ce7/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"09b66910f8d7d6f8f3eb1dc82d6395ad6aca7e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 23:20:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d5b360bc02bb8ca4d37be307ff9a109721ddd867","revisions":{"9d99d6e48967849dd5e0447e6a4d335025d2cd88":{"kind":"REWORK","_number":1,"created":"2019-04-29 21:58:22.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/11/656411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/11/656411/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/1"}}},"commit":{"parents":[{"commit":"24a7cb633f697e7e8a11055fd89b467e1d8bc993","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24a7cb633f697e7e8a11055fd89b467e1d8bc993"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-29 21:57:40.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-29 21:57:40.000000000","tz":-180},"subject":"Added rewriting rules","message":"Added rewriting rules\n\nadded rules to redirect registration/password lost\nand resend verification to www site\n\nChange-Id: I487a973826fb40675e5e9999be89d69481a7312e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/9d99d6e48967849dd5e0447e6a4d335025d2cd88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/9d99d6e48967849dd5e0447e6a4d335025d2cd88"}]},"branch":"refs/heads/master"},"1cb8a80d59654d8948cb0bc9ef42d420ef100cf7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-29 21:59:02.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/11/656411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/11/656411/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/2"}}},"commit":{"parents":[{"commit":"24a7cb633f697e7e8a11055fd89b467e1d8bc993","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24a7cb633f697e7e8a11055fd89b467e1d8bc993"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-29 21:57:40.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-29 21:58:27.000000000","tz":-180},"subject":"Added rewriting rules","message":"Added rewriting rules\n\nadded rules to redirect registration/password lost\nand resend verification to www site\n\nChange-Id: I487a973826fb40675e5e9999be89d69481a7312e\nDepends-On: https://review.opendev.org/#/c/656396/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1cb8a80d59654d8948cb0bc9ef42d420ef100cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1cb8a80d59654d8948cb0bc9ef42d420ef100cf7"}]},"branch":"refs/heads/master"},"d5b360bc02bb8ca4d37be307ff9a109721ddd867":{"kind":"REWORK","_number":3,"created":"2019-04-29 22:02:28.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/11/656411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/11/656411/3","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/11/656411/3"}}},"commit":{"parents":[{"commit":"24a7cb633f697e7e8a11055fd89b467e1d8bc993","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24a7cb633f697e7e8a11055fd89b467e1d8bc993"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-29 21:57:40.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-29 22:02:17.000000000","tz":-180},"subject":"Added rewriting rules","message":"Added rewriting rules\n\nadded rules to redirect registration/password lost\nand resend verification to www site\n\nChange-Id: I487a973826fb40675e5e9999be89d69481a7312e\nDepends-On: https://review.opendev.org/#/c/656396/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/d5b360bc02bb8ca4d37be307ff9a109721ddd867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/d5b360bc02bb8ca4d37be307ff9a109721ddd867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
