)]}'
{"id":"opendev%2Fpuppet-openstackid~656427","triplet_id":"opendev%2Fpuppet-openstackid~master~I705223d3b7c80cf482149b5eafa02a682cc8f5a6","project":"opendev/puppet-openstackid","branch":"master","topic":"hotfix/redirect_urls","hashtags":[],"change_id":"I705223d3b7c80cf482149b5eafa02a682cc8f5a6","subject":"Fix on redirect urls regex","status":"MERGED","created":"2019-04-30 01:15:02.000000000","updated":"2019-04-30 05:21:25.000000000","submitted":"2019-04-30 05:21:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656427-1556601685878-82ae42cf","meta_rev_id":"3e53472f0a8a189d902b734abd4e71558f68f9c8","_number":656427,"virtual_id_number":656427,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-04-30 05:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-04-30 02:06:38.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-30 04:55:16.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-30 04:55:16.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 02:06:38.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-30 04:55:16.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-04-30 05:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"934f2715a959e98ef444748ba52b3f52a8443551","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2019-04-30 01:15:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37bdaaa4604239eb3387f305ba3456c1e52d34ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 01:41:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/27/656427/1/check/legacy-puppet-lint/527dafd/ : SUCCESS in 5m 01s\n- legacy-puppet-syntax-3 http://logs.openstack.org/27/656427/1/check/legacy-puppet-syntax-3/00d64dd/ : SUCCESS in 8m 42s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/27/656427/1/check/legacy-puppet-syntax-4-centos-7/a3273eb/ : SUCCESS in 12m 38s\n- infra-puppet-apply-3-ubuntu-trusty http://logs.openstack.org/27/656427/1/check/infra-puppet-apply-3-ubuntu-trusty/28dcd79/ : SUCCESS in 5m 26s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/27/656427/1/check/infra-puppet-apply-3-ubuntu-xenial/6b1ea4b/ : SUCCESS in 7m 52s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/27/656427/1/check/infra-puppet-apply-4-ubuntu-xenial/d0cd30b/ : SUCCESS in 13m 59s\n- openstackci-beaker-ubuntu-trusty http://logs.openstack.org/27/656427/1/check/openstackci-beaker-ubuntu-trusty/9f91980/ : SUCCESS in 10m 08s\n- openstackci-beaker http://logs.openstack.org/27/656427/1/check/openstackci-beaker/f3d5d22/ : SUCCESS in 16m 09s\n- openstackci-beaker-puppet-4 http://logs.openstack.org/27/656427/1/check/openstackci-beaker-puppet-4/16e2128/ : SUCCESS in 17m 31s (non-voting)\n- puppet-beaker-rspec-centos-7-infra http://logs.openstack.org/27/656427/1/check/puppet-beaker-rspec-centos-7-infra/a847229/ : SUCCESS in 9m 36s\n- puppet-beaker-rspec-puppet-4-centos-7-infra http://logs.openstack.org/27/656427/1/check/puppet-beaker-rspec-puppet-4-centos-7-infra/d9b6f19/ : SUCCESS in 12m 18s (non-voting)\n- puppet-beaker-rspec-infra http://logs.openstack.org/27/656427/1/check/puppet-beaker-rspec-infra/80576e0/ : SUCCESS in 4m 36s\n- puppet-beaker-rspec-puppet-4-infra http://logs.openstack.org/27/656427/1/check/puppet-beaker-rspec-puppet-4-infra/8924c0a/ : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27d515e17d1f123224c937db808eb3ddc46f081a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-30 02:06:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12b4c954d771ccbb957ea64707f515fa14a239fa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-04-30 04:55:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4630bb819fd73fe8526cc80f97e43ff17ee38eef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 04:55:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2446299f499a074b817007e7e5e6857fbbe92d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 05:21:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa757700486e44074dd3179800b286415bcccb0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 05:21:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint http://logs.openstack.org/27/656427/1/gate/legacy-puppet-lint/e99966b/ : SUCCESS in 3m 21s\n- legacy-puppet-syntax-3 http://logs.openstack.org/27/656427/1/gate/legacy-puppet-syntax-3/c5a4b13/ : SUCCESS in 5m 43s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/27/656427/1/gate/legacy-puppet-syntax-4-centos-7/b4486f3/ : SUCCESS in 7m 32s\n- infra-puppet-apply-3-ubuntu-trusty http://logs.openstack.org/27/656427/1/gate/infra-puppet-apply-3-ubuntu-trusty/a5ffb51/ : SUCCESS in 6m 53s\n- infra-puppet-apply-3-ubuntu-xenial http://logs.openstack.org/27/656427/1/gate/infra-puppet-apply-3-ubuntu-xenial/53ac98a/ : SUCCESS in 8m 27s\n- openstackci-beaker-ubuntu-trusty http://logs.openstack.org/27/656427/1/gate/openstackci-beaker-ubuntu-trusty/5c623f1/ : SUCCESS in 10m 57s\n- infra-puppet-apply-4-ubuntu-xenial http://logs.openstack.org/27/656427/1/gate/infra-puppet-apply-4-ubuntu-xenial/376064e/ : SUCCESS in 13m 45s\n- openstackci-beaker http://logs.openstack.org/27/656427/1/gate/openstackci-beaker/2971645/ : SUCCESS in 18m 07s\n- puppet-beaker-rspec-centos-7-infra http://logs.openstack.org/27/656427/1/gate/puppet-beaker-rspec-centos-7-infra/f0fefed/ : SUCCESS in 12m 04s\n- puppet-beaker-rspec-infra http://logs.openstack.org/27/656427/1/gate/puppet-beaker-rspec-infra/a932363/ : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83c937e34af218aafa9de4549f3d6b9800563444","revisions":{"83c937e34af218aafa9de4549f3d6b9800563444":{"kind":"REWORK","_number":1,"created":"2019-04-30 01:15:02.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/27/656427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/27/656427/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/27/656427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/27/656427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/27/656427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/27/656427/1"}}},"commit":{"parents":[{"commit":"1a8fd41b3e749db701f17a4e5a57f110498ab4d9","subject":"Merge \"Added rewriting rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1a8fd41b3e749db701f17a4e5a57f110498ab4d9"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-30 01:14:35.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2019-04-30 01:14:35.000000000","tz":-180},"subject":"Fix on redirect urls regex","message":"Fix on redirect urls regex\n\nChange-Id: I705223d3b7c80cf482149b5eafa02a682cc8f5a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/83c937e34af218aafa9de4549f3d6b9800563444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/83c937e34af218aafa9de4549f3d6b9800563444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
