)]}'
{"id":"opendev%2Fpuppet-openstackid~703720","triplet_id":"opendev%2Fpuppet-openstackid~master~I633add5af8d96223d331a129f39956b1af4f8867","project":"opendev/puppet-openstackid","branch":"master","topic":"feature/idp_user_management_migration","hashtags":[],"change_id":"I633add5af8d96223d331a129f39956b1af4f8867","subject":"Updated build","status":"MERGED","created":"2020-01-22 03:15:17.000000000","updated":"2020-01-22 05:57:09.000000000","submitted":"2020-01-22 05:57:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703720-1579672629304-090dce9f","meta_rev_id":"ed7ab6ac9ffe3790ae1bef417f6b8fe7b82e1dbd","_number":703720,"virtual_id_number":703720,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 05:57:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-22 04:57:12.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2020-01-22 04:36:37.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2020-01-22 05:39:39.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 04:57:12.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-01-22 05:39:39.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-22 05:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d21ff51a21e6960cc75b1ac59ff00cb82ce430a","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-22 03:15:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e23087e3be3465d8f87e3ef3bad38f057dce0a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 03:33:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/17338c1e73b149b2bdbe2dca88407517 : FAILURE in 5m 08s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8dd642e4b59d4bbcb4eeb56d4ed228b1 : SUCCESS in 10m 54s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/90fdad8d3ad84b4194ab8d3517ef5352 : SUCCESS in 16m 04s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/a7cfc3a422074d518682efcacadd974a : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4c7d199449a3eb9526383bbb7a84990b63996673","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 03:52:09.000000000","message":"Patch Set 1:\n\nLooks like this module has fallen behind recent changes to the linter:\n\n    manifests/init.pp - WARNING: mode should be represented as a 4 digit octal value or symbolic mode on line 344\n    manifests/init.pp - WARNING: class included by absolute name (::$class) on line 307\n    manifests/init.pp - WARNING: class included by absolute name (::$class) on line 432\n    manifests/init.pp - WARNING: class included by absolute name (::$class) on line 438\n    manifests/init.pp - WARNING: class included by absolute name (::$class) on line 457\n    manifests/init.pp - WARNING: class included by absolute name (::$class) on line 458\n    manifests/init.pp - WARNING: class included by absolute name (::$class) on line 459\n\nI\u0027ll quickly push up a change to fix these problems and rebase yours on it.","accounts_in_message":[],"_revision_number":1},{"id":"8bd5f265f0c57a08ef42ee0e57615a1a61866d2e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 03:58:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\nI\u0027ll take care of the correction marked inline when I rebase this.","accounts_in_message":[],"_revision_number":1},{"id":"77caad7af5e45d295a51313012463efc30adbd6a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 04:07:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ff7e84b82c92c5d028f3e9539958a5048b58812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 04:30:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/cef34afc817b44b09ba1cd50ddcaae6c : SUCCESS in 5m 02s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/13216d305aa94d1ca4173418a343fa75 : SUCCESS in 16m 32s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/d422bfd068a945ffac6eca3904c21f8e : SUCCESS in 19m 23s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/2d5fdd23560446099cc3a847e4b85980 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8d7c0353b82ed916016299cbf97a72d22c239fed","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 04:36:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cbb64d69c664327df9ba58843b099d677ce888d9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-22 04:57:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"742890e3ca73314c020e3e1ee4cdc4c128f98e7a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 05:39:39.000000000","message":"Patch Set 2: Workflow+1\n\nIt\u0027s safe to approve anyway since I put the production server in our emergency disable list so that only the dev server is getting this update for now until it\u0027s confirmed working.","accounts_in_message":[],"_revision_number":2},{"id":"546e17373ffbf960be97917cc3b953fcaed3a6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:39:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f57fb62841e48d9d3599e8e8b8a66feae12f9fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:57:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/aa8ab07bbb004397a47417b5fad9515d : SUCCESS in 4m 44s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6fc2a704cb6c457c833b08f0e988dcb1 : SUCCESS in 11m 26s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/30629ec1c83a41219d401db26e13b76b : SUCCESS in 11m 24s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/98d374935cb44094a97619010dbfe734 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6747283fddd7492cde16bf2671c7f18a99f7969b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:57:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24076bd286884db72774fc60d1566e4175602503","revisions":{"3e9ffa611db33e8a1c23d593a220399c40536881":{"kind":"REWORK","_number":1,"created":"2020-01-22 03:15:17.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/20/703720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/20/703720/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/1"}}},"commit":{"parents":[{"commit":"8341bbfd326a35e9dc34fae77f95dd6b8983cc53","subject":"Added missing double quotes for some config variables on .env file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/8341bbfd326a35e9dc34fae77f95dd6b8983cc53"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 03:14:34.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 03:14:34.000000000","tz":-180},"subject":"Updated build","message":"Updated build\n\nUpdated puppet build to support new migration\nfor user management from IDP side\n\nChange-Id: I633add5af8d96223d331a129f39956b1af4f8867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/3e9ffa611db33e8a1c23d593a220399c40536881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/3e9ffa611db33e8a1c23d593a220399c40536881"}]},"branch":"refs/heads/master"},"24076bd286884db72774fc60d1566e4175602503":{"kind":"REWORK","_number":2,"created":"2020-01-22 04:07:28.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/20/703720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/20/703720/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/20/703720/2"}}},"commit":{"parents":[{"commit":"7091e5ced9bd2e90b9090ae05ad64c12878011ce","subject":"Puppet-lint says \"class included by absolute name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/7091e5ced9bd2e90b9090ae05ad64c12878011ce"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 03:14:34.000000000","tz":-180},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-01-22 04:07:17.000000000","tz":0},"subject":"Updated build","message":"Updated build\n\nUpdated puppet build to support new migration\nfor user management from IDP side\n\nChange-Id: I633add5af8d96223d331a129f39956b1af4f8867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24076bd286884db72774fc60d1566e4175602503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24076bd286884db72774fc60d1566e4175602503"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
