)]}'
{"id":"opendev%2Fpuppet-openstackid~703834","triplet_id":"opendev%2Fpuppet-openstackid~master~I65099e2ee2a0b3b153d70cb9ba6b7e96ec9baced","project":"opendev/puppet-openstackid","branch":"master","topic":"hotfix/supervisord","hashtags":[],"change_id":"I65099e2ee2a0b3b153d70cb9ba6b7e96ec9baced","subject":"Updated Build","status":"MERGED","created":"2020-01-22 16:25:52.000000000","updated":"2020-01-24 15:44:01.000000000","submitted":"2020-01-24 15:44:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703834-1579880641229-bbef5f2b","meta_rev_id":"812f27781069b1956a260fa3c9fb46e6f4c3a88f","_number":703834,"virtual_id_number":703834,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 15:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":2,"date":"2020-01-24 14:53:59.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-24 15:00:26.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-24 15:00:26.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 16:29:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-22 16:29:12.000000000","updated_by":{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},"reviewer":{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},"state":"REVIEWER"},{"updated":"2020-01-24 14:53:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-01-24 15:00:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-24 15:44:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7023f1270bf303779efd3cbf60cfb0ed2b1a47c6","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-22 16:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb7b7f8627de9fa9f38524b30adbc1db30611c13","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-22 16:29:21.000000000","message":"Patch Set 1: Code-Review+1\n\nplease review :)","accounts_in_message":[],"_revision_number":1},{"id":"251ec486d37d93201ff00fd217a68316fbf62171","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 17:04:40.000000000","message":"Patch Set 1:\n\nI don\u0027t see any fix to the supervisor launch script, which you mention in the commit message. Did you forget to include it in the patch?\n\nAs for the docroot change, it seems a little strange to have the variable named \"docroot\" not represent the actual docroot value for the vhost. Can you explain why you needed to change that?","accounts_in_message":[],"_revision_number":1},{"id":"2877c7135160d4ee890efe73e8a1b84d26decce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/a1bebe22c54a41979ba5ae45d8cc423c : SUCCESS in 8m 17s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7414cf6a271d47b69fd5f68407131602 : SUCCESS in 12m 33s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/2b09be1216a24c5c80a34022e75c8e4e : SUCCESS in 43m 36s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/6a9d4f958b724a43aa0a82c8432aef18 : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":1},{"id":"14cf20f0104200c58c26030c9b7ddb9d75e16e51","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-22 17:45:42.000000000","message":"Patch Set 1: -Code-Review\n\nJeremy the docroot in my POV is srv/openstackid/w not srv/openstackid/w/public\n\nas was set\nor i could create another variable called projectroot\u003dsrv/openstackid/w and set doctroot to $projectroot/public, reference to supervisor was not included bc its already using docroot variable ( that is suposed to be pointing to  srv/openstackid/w)\n\nregards","accounts_in_message":[],"_revision_number":1},{"id":"fa6bb85152bc0b29b31c6311ef0786d6e5013541","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-22 17:52:33.000000000","message":"Patch Set 1:\n\nBut the vhost template is setting \"DocumentRoot \u003c%\u003d @docroot %\u003e/public\" which is... well... somewhat confusing.","accounts_in_message":[],"_revision_number":1},{"id":"17a78d7966ac96cc3e41c87191d8f94b60e31ac6","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-22 17:56:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"741d3ae01aa368050bcd60cac94eba356c985514","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-22 18:03:37.000000000","message":"Patch Set 2:\n\nOK i created a new variable projectroot\ni do think that is more clear :)\nregards","accounts_in_message":[],"_revision_number":2},{"id":"9de2cb45dc16bb66fd4e565ace4748501e750a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 18:28:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/3372ed99faa14bc499db9414e1b4a514 : SUCCESS in 7m 05s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/297d5dc2cf734988ab27b05080a0cec8 : SUCCESS in 23m 12s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/00871cb427564bb0880d50c3ab89f4a4 : SUCCESS in 16m 41s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/d517c18976d046649f253cba5ac203a6 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"dcde492467d76327e9154cb7e3f2e218c9085856","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-23 00:40:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3df804dc5fb249729553d5fbf8354683d68cebb5","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-24 14:11:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5dad351e936d4aa5cba80c26ea8eef50ee51db34","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-24 14:12:25.000000000","message":"Patch Set 3:\n\ncould we approve this one so i could start testing the build at dev ? :)\nregards","accounts_in_message":[],"_revision_number":3},{"id":"cc6635bb91ce65887e5fd353a89ea5bfea1d90cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 14:42:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/32617582719547c9bb43d5f33ce799e4 : FAILURE in 7m 07s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7e24d4604e794c649bb0c067f86f1145 : SUCCESS in 20m 47s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/c0cd5dc9895d4f42b873796e1ed7ce48 : SUCCESS in 14m 30s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/4ef1999cf34a4119b4f8700423184571 : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":3},{"id":"eea22ec306e59e522a7de5939db2d9f14209f25d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-24 14:47:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb26ea6b337873c4efe6f47b9d29308c35e0ac08","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-24 14:52:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6018ca9fb940433f1d2d60d0645c6670091e9113","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-24 14:53:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6f1d6e2060a9b87af0c9138e8c7bec42e0367b9d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-24 15:00:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"68a296d5afafce731a5f0c324d379c71b075f0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:19:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/c081def215b242ed8f4a44a256a9a76b : SUCCESS in 6m 02s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/12ba2135a2414eb0a1b50beec1aadb79 : SUCCESS in 21m 37s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/9d44d108bf424e1191afb3727db6eab5 : SUCCESS in 15m 13s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/7c0c0b45fe604b8389c0ab86ecfb3ce5 : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":4},{"id":"a913d53d347eae7daad01158544cd670dc43b613","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:19:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"07a282ffa755d32b7624f939941d2a5466d139a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:44:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/64fc60a87db145c9820600ec89a99032 : SUCCESS in 6m 03s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/57ce78ed7ef2411496bf03398c5e30cc : SUCCESS in 13m 36s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/4151927f7cba4e2aa7ed641577b01b62 : SUCCESS in 15m 34s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/675d6e5318d24ba2b93924ed54d822ab : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":4},{"id":"a81fe5b1e12454c22a64123f9f954ce4caffed2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:44:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1806356c5242f66dccc840fdb7fae22f2f8226f9","revisions":{"2b95bd578c4ef61e28065d387d55901c2bbc6bb4":{"kind":"REWORK","_number":1,"created":"2020-01-22 16:25:52.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/34/703834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/34/703834/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/1"}}},"commit":{"parents":[{"commit":"24076bd286884db72774fc60d1566e4175602503","subject":"Updated build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24076bd286884db72774fc60d1566e4175602503"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 16:25:10.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 16:25:10.000000000","tz":-180},"subject":"Updated Build","message":"Updated Build\n\n* Fix on supervisor launch script\n* Refactoring on docroot variable\n\nChange-Id: I65099e2ee2a0b3b153d70cb9ba6b7e96ec9baced\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/2b95bd578c4ef61e28065d387d55901c2bbc6bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/2b95bd578c4ef61e28065d387d55901c2bbc6bb4"}]},"branch":"refs/heads/master"},"5a7a79762ce944a016212e18afbbf874b5edd786":{"kind":"REWORK","_number":2,"created":"2020-01-22 17:56:23.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/34/703834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/34/703834/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/2"}}},"commit":{"parents":[{"commit":"24076bd286884db72774fc60d1566e4175602503","subject":"Updated build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24076bd286884db72774fc60d1566e4175602503"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 16:25:10.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 17:56:12.000000000","tz":-180},"subject":"Updated Build","message":"Updated Build\n\n* Fix on supervisor launch script\n* Refactoring on docroot variable\n\nChange-Id: I65099e2ee2a0b3b153d70cb9ba6b7e96ec9baced\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/5a7a79762ce944a016212e18afbbf874b5edd786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/5a7a79762ce944a016212e18afbbf874b5edd786"}]},"branch":"refs/heads/master"},"82fa56db283c894f73be4dad56623b881930e37b":{"kind":"REWORK","_number":3,"created":"2020-01-24 14:11:42.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/34/703834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/34/703834/3","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/3"}}},"commit":{"parents":[{"commit":"24076bd286884db72774fc60d1566e4175602503","subject":"Updated build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24076bd286884db72774fc60d1566e4175602503"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 16:25:10.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-24 14:11:33.000000000","tz":-180},"subject":"Updated Build","message":"Updated Build\n\n* Fix on supervisor launch script\n* Refactoring on docroot variable\n\nChange-Id: I65099e2ee2a0b3b153d70cb9ba6b7e96ec9baced\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/82fa56db283c894f73be4dad56623b881930e37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/82fa56db283c894f73be4dad56623b881930e37b"}]},"branch":"refs/heads/master"},"1806356c5242f66dccc840fdb7fae22f2f8226f9":{"kind":"REWORK","_number":4,"created":"2020-01-24 14:52:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/34/703834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/34/703834/4","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/34/703834/4"}}},"commit":{"parents":[{"commit":"24076bd286884db72774fc60d1566e4175602503","subject":"Updated build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/24076bd286884db72774fc60d1566e4175602503"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-22 16:25:10.000000000","tz":-180},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-01-24 14:51:58.000000000","tz":0},"subject":"Updated Build","message":"Updated Build\n\n* Fix on supervisor launch script\n* Refactoring on docroot variable\n\nChange-Id: I65099e2ee2a0b3b153d70cb9ba6b7e96ec9baced\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1806356c5242f66dccc840fdb7fae22f2f8226f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1806356c5242f66dccc840fdb7fae22f2f8226f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
