)]}'
{"id":"opendev%2Fpuppet-openstackid~704329","triplet_id":"opendev%2Fpuppet-openstackid~master~Icdb4740621d6f9c7310d1b6bb52ce1db43652882","project":"opendev/puppet-openstackid","branch":"master","topic":"hotfix/artisan_permissions","hashtags":[],"change_id":"Icdb4740621d6f9c7310d1b6bb52ce1db43652882","subject":"Refactored Deployment Script Functions","status":"MERGED","created":"2020-01-27 14:35:00.000000000","updated":"2020-01-29 23:05:20.000000000","submitted":"2020-01-29 23:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704329-1580339120910-46a14a57","meta_rev_id":"da0e8b355a55586e81ca7e1fb706edf4266c04c7","_number":704329,"virtual_id_number":704329,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 23:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 22:04:24.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-29 22:04:24.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 18:18:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-29 21:58:28.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-29 22:04:24.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-29 23:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf581c4c6eee30a7bd7ded3b04d8c430f54bdbea","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-27 14:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46052bdea532111a60602c598cb484ec8f37399f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/4d8e3a4f6a8244e8bb5df7cd9b5ef9d8 : SUCCESS in 6m 14s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/32a876ce91af496c9543c8c5e3566342 : SUCCESS in 13m 37s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/447656bdc6304891ac13afd8c89ff5a4 : SUCCESS in 14m 07s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/dac40099e3774f58a0bef55bf0b3a77f : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cedbaf0920462ac27280097c194221ab8e25bdc0","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-28 14:43:28.000000000","message":"Patch Set 1: Code-Review+1\n\nplease review regards","accounts_in_message":[],"_revision_number":1},{"id":"b3d2803b1aed1649cc831e3d72b72b81f08471b2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-28 17:11:59.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c81c20e63d0cdcabac9f1136229f0a2c1dfffb96","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-28 17:23:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35f7c97d4ca007cb1aecbb28b4be3887bd7e6d80","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-28 17:28:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27ee62598b03742b970edf70a21d6764618feedc","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-28 17:30:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"90845c65b9bb39f7a847870ef6bea2ff3cb1d294","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-28 17:38:09.000000000","message":"Patch Set 2: Code-Review+1\n\nupdated commit message and refactored a little more also corrected your suggestions\ncheers","accounts_in_message":[],"_revision_number":2},{"id":"1e8e70870e7ef6ff4c4107a75cd5cc7f5e0d1a59","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-28 17:55:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d7d9daa6cd92179a933ec99f6d193c276365d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:57:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/a5d7995557a9482484da313a245f42f1 : SUCCESS in 5m 20s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/915be408ff0d41d09c0baf04ec756276 : SUCCESS in 15m 09s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/2f49f04187be4a5a8391876448fde7d5 : SUCCESS in 19m 25s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/23a474737ca248b2b47ec9839053cbf2 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6dfb12dec0c70ea28e50a488364cfc72fc2354c4","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-29 21:53:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38555de16263ce4a1f0ef298e8dbdc28c164c72c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-29 21:58:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"54b09d4fde359c3f02d97420cb0672ae9386c118","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-01-29 22:03:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa00b0a66237895de199a0a70b909a4728b01e4f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-29 22:04:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"418fc10d478bdc36b8bfd23bd75dce484fdd0ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:30:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/17a420ea31c94bdd8401f9c18e2d1a30 : SUCCESS in 8m 15s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ae00608f55634d708ca3479ecd25b150 : SUCCESS in 14m 30s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/ddb537db5bca46c38eba452d22bbc294 : SUCCESS in 22m 50s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/3e3ce2fdf2334a66b4b0059a2b6b72d5 : SUCCESS in 16m 50s","accounts_in_message":[],"_revision_number":4},{"id":"1dcd0774f16cd90f066de6b4dc75d1076c7d7177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:30:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"56d638461ba19b7a7d401cc75ecc07f4c119112f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:05:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/40ac8f68540f49788d00552e0857eba3 : SUCCESS in 7m 04s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7c388df3b1dd49fb91d5dc1b3846438e : SUCCESS in 21m 45s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/3bcd9afc830e4be6937be2a9d5464563 : SUCCESS in 21m 22s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/9a4a75ccdcd5445b8c803fcab004f990 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":4},{"id":"fcd81aac05a3b40f0c1ef809ed72f65d78c26639","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2cd724f1aafbafa5f81799ae6c0ef6802d9ea807","revisions":{"22433bb0efd0ca2c6c7c040bc7709e72f121258d":{"kind":"REWORK","_number":1,"created":"2020-01-27 14:35:00.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/29/704329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/29/704329/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/1"}}},"commit":{"parents":[{"commit":"1806356c5242f66dccc840fdb7fae22f2f8226f9","subject":"Updated Build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1806356c5242f66dccc840fdb7fae22f2f8226f9"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-27 14:34:19.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-27 14:34:19.000000000","tz":-180},"subject":"Fixed artisan command permissions","message":"Fixed artisan command permissions\n\nSet proper execution permissions for artisan command\n(www-data)\n\nChange-Id: Icdb4740621d6f9c7310d1b6bb52ce1db43652882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/22433bb0efd0ca2c6c7c040bc7709e72f121258d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/22433bb0efd0ca2c6c7c040bc7709e72f121258d"}]},"branch":"refs/heads/master"},"a06d2eeffa609d1a2f39a6f4ed81a90265049413":{"kind":"REWORK","_number":2,"created":"2020-01-28 17:28:12.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/29/704329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/29/704329/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/2"}}},"commit":{"parents":[{"commit":"1806356c5242f66dccc840fdb7fae22f2f8226f9","subject":"Updated Build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1806356c5242f66dccc840fdb7fae22f2f8226f9"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-27 14:34:19.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-28 17:26:52.000000000","tz":-180},"subject":"Refactored Deployment Script Functions","message":"Refactored Deployment Script Functions\n\n* Moved Redis clear cache routine to its own function.\n* Moved ORM clear cache routine to its own function.\n* Set proper execution permissions for artisan command.\n\nChange-Id: Icdb4740621d6f9c7310d1b6bb52ce1db43652882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/a06d2eeffa609d1a2f39a6f4ed81a90265049413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/a06d2eeffa609d1a2f39a6f4ed81a90265049413"}]},"branch":"refs/heads/master"},"f1c32c518caa42b85d3859abf2ff5448cdefc237":{"kind":"REWORK","_number":3,"created":"2020-01-29 21:53:48.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/29/704329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/29/704329/3","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/3"}}},"commit":{"parents":[{"commit":"1806356c5242f66dccc840fdb7fae22f2f8226f9","subject":"Updated Build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1806356c5242f66dccc840fdb7fae22f2f8226f9"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-27 14:34:19.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-29 21:53:37.000000000","tz":-180},"subject":"Refactored Deployment Script Functions","message":"Refactored Deployment Script Functions\n\n* Moved Redis clear cache routine to its own function.\n* Moved ORM clear cache routine to its own function.\n* Set proper execution permissions for artisan command.\n\nChange-Id: Icdb4740621d6f9c7310d1b6bb52ce1db43652882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/f1c32c518caa42b85d3859abf2ff5448cdefc237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/f1c32c518caa42b85d3859abf2ff5448cdefc237"}]},"branch":"refs/heads/master"},"2cd724f1aafbafa5f81799ae6c0ef6802d9ea807":{"kind":"REWORK","_number":4,"created":"2020-01-29 22:03:13.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/29/704329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/29/704329/4","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/29/704329/4"}}},"commit":{"parents":[{"commit":"1806356c5242f66dccc840fdb7fae22f2f8226f9","subject":"Updated Build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/1806356c5242f66dccc840fdb7fae22f2f8226f9"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-27 14:34:19.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-01-29 22:03:00.000000000","tz":-180},"subject":"Refactored Deployment Script Functions","message":"Refactored Deployment Script Functions\n\n* Moved Redis clear cache routine to its own function.\n* Moved ORM clear cache routine to its own function.\n* Set proper execution permissions for artisan command.\n\nChange-Id: Icdb4740621d6f9c7310d1b6bb52ce1db43652882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/2cd724f1aafbafa5f81799ae6c0ef6802d9ea807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/2cd724f1aafbafa5f81799ae6c0ef6802d9ea807"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
