)]}'
{"id":"opendev%2Fpuppet-openstackid~708131","triplet_id":"opendev%2Fpuppet-openstackid~master~I75aca729ebaa9b3bf6c367592b0c42dd9221a856","project":"opendev/puppet-openstackid","branch":"master","topic":"hotfix/deployment_script_refactoring","hashtags":[],"change_id":"I75aca729ebaa9b3bf6c367592b0c42dd9221a856","subject":"Updated deployment scripts","status":"MERGED","created":"2020-02-17 14:43:51.000000000","updated":"2020-02-17 16:08:13.000000000","submitted":"2020-02-17 16:08:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708131-1581955693509-798fc2db","meta_rev_id":"33fe393874dd2e05b1857b8ba17b2ad1aca392b3","_number":708131,"virtual_id_number":708131,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 16:08:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-02-17 14:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-17 15:44:31.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"date":"2020-02-17 14:45:24.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"date":"2020-02-17 14:45:53.000000000","_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-17 15:44:31.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 14:45:24.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-17 14:45:53.000000000","updated_by":{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},"reviewer":{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},"state":"REVIEWER"},{"updated":"2020-02-17 15:44:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-02-17 16:08:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03f9fa8a38543a43ece4da52c86604fc78ce341b","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-17 14:43:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d358f6f4a387a9b5134ed7226d41afc6897335d2","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-17 14:46:13.000000000","message":"Patch Set 1: Code-Review+1\n\nplease review, this is inline with the efforts to move the new idp code to production\nregards","accounts_in_message":[],"_revision_number":1},{"id":"f1359c93b9ba4de23a3b88a966ccb59a1e4bb408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:05:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/f23bf8cac42b4f1f90b153ed8a86bb15 : SUCCESS in 5m 39s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/52eb07e579c8480b8aa1d2ad56cd4267 : SUCCESS in 13m 22s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/4edfce23d81a4c6c9556d566a2f4b709 : SUCCESS in 15m 33s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/ee9491d7adda4ee897354688038efcfd : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8be6d8c3bc8c1839a13785b9776a1f7df1534357","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-17 15:44:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)\n\nMost of the patch is just drying up code duplication by relocating into a named function, but I\u0027ve made a couple comments on the relocated code if you decide you want to simplify it while you\u0027re in this part of the automation.","accounts_in_message":[],"_revision_number":1},{"id":"5bf1dd1fdfd4eeeb35335fa02b90eef2268291dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:44:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee86b0f73fc6c91eb3c10e49d9a6107fd373f347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:08:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d2c1e20aa38516a9ece797b98a6cd8299dcc99cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:08:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/215cbb6167584bd488ebb92a1d7cd26f : SUCCESS in 6m 02s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6e06bc0cfd1f4cb9b375fcd06fc4af40 : SUCCESS in 13m 28s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/5d048acc8c544fa1a03720ccff1ecc26 : SUCCESS in 17m 36s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/c223e4b1dd4a4639870cfc59b587fff8 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb0da02f2c0f25c4665553472a17dbafe1a42ea8","revisions":{"fb0da02f2c0f25c4665553472a17dbafe1a42ea8":{"kind":"REWORK","_number":1,"created":"2020-02-17 14:43:51.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/31/708131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/31/708131/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/31/708131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/31/708131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/31/708131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/31/708131/1"}}},"commit":{"parents":[{"commit":"d135337253b69185cc0187b2e0bffda06b22e1f2","subject":"Fixed supervisor watchdog regex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/d135337253b69185cc0187b2e0bffda06b22e1f2"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-02-17 14:43:10.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-02-17 14:43:10.000000000","tz":-180},"subject":"Updated deployment scripts","message":"Updated deployment scripts\n\n* setting the proper storage permissions\n  at the end of deployment\n* refactoring\n\nChange-Id: I75aca729ebaa9b3bf6c367592b0c42dd9221a856\nSigned-off-by: smarcet \u003csmarcet@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/fb0da02f2c0f25c4665553472a17dbafe1a42ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/fb0da02f2c0f25c4665553472a17dbafe1a42ea8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
