)]}'
{"id":"opendev%2Fpuppet-openstackid~709525","triplet_id":"opendev%2Fpuppet-openstackid~master~Id57bbf6329a0aac9904fd317ce0e1d8628993aa3","project":"opendev/puppet-openstackid","branch":"master","topic":"feature/removed_stale_variables","hashtags":[],"change_id":"Id57bbf6329a0aac9904fd317ce0e1d8628993aa3","subject":"Removed stale config variables","status":"MERGED","created":"2020-02-24 13:19:42.000000000","updated":"2020-02-24 22:29:47.000000000","submitted":"2020-02-24 22:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709525-1582583387322-577aa814","meta_rev_id":"939d60ae02a565602f11e0f690188d80d5171cb5","_number":709525,"virtual_id_number":709525,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 22:29:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-24 19:13:03.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-24 21:58:47.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 13:29:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-24 13:29:09.000000000","updated_by":{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},"reviewer":{"_account_id":10278,"name":"Jimmy McArthur","email":"jimmy@openstack.org","username":"jimmymcarthur"},"state":"REVIEWER"},{"updated":"2020-02-24 21:58:47.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-02-24 22:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8a0169b3f471b63bff393c32db62ce4aa1210ce","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-24 13:19:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74101ef198975901905acdd95cb2290e6f4e01b1","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-24 13:28:42.000000000","message":"Patch Set 1: Code-Review+1\n\ni removed silverstripe related config variables due the idp migration to use an unique user store\nwe are planning to deploy all news changes ( if possible ) today to production\nregards","accounts_in_message":[],"_revision_number":1},{"id":"058df93996b41a3997a28894a97ef62ac9b1bf58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:09:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/090ff49c34704fdf99cded9dd7e57096 : SUCCESS in 6m 54s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/448672a1c9f04767b223a4fb50035db5 : FAILURE in 12m 15s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/eb85967cec5842549d893a54bd40ff41 : SUCCESS in 14m 09s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/b32d3942335645218d28e22df935922c : SUCCESS in 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2ab1b0ea567aa57528ea5739f892ec70acfa8c1e","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-24 14:28:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"512c11e791b3e46d1401659c96379d227373dc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:57:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/3b194140784c41f6bf3b32b109595c2b : SUCCESS in 7m 00s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/eaa597a42c2f4666bf32686004eeb243 : FAILURE in 18m 06s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/8977abfd26c74033abb37e2c64b9f351 : SUCCESS in 20m 39s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/b297facd2c144c86a53feaf0e1a516cf : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"190dfaf99aaca9a51d62a4a28b9923479359c697","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-24 17:29:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1fb9ac1161d66b82e7cb6001da6136b56b2ca201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:58:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/76157d0571fc415ea679972c78717aa7 : SUCCESS in 5m 26s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0b97aa33d43e4a0796724ded38ad126c : FAILURE in 13m 08s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/586ab59d980c4643b434b9950f2eaef9 : SUCCESS in 17m 26s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/7d1b1667b1824abc91c541c8da4c676e : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"385647fadba79ca89436dcfb7749ce7d9b61a957","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-24 18:31:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f67955d5862c4a7c9297f67b0c8a75073a5718e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:02:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/37eddff757514a4e8caee91d4838cca7 : SUCCESS in 8m 19s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/957a13903a99469c8628b69921435a6a : SUCCESS in 21m 04s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/389f6aea5aef4a5d89fc86785269d74a : SUCCESS in 23m 06s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/843293ba8bb54b44bac10a70d23d9397 : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3d861676338b14e8fa4c81345ac0e6208809cd23","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-24 19:13:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbe65b9a043d303a992a5d6307be87cc3724face","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-24 21:58:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fee59bf50af45dcae7339e1faaa549b63d8ac9b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 21:59:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eeae409953365b5771dd25b37b9602b35b5b593b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 22:29:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5e43b36193dd36c6778675ccdf3b51021cf50779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 22:29:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/1ac8014462424bc68cbbea3de365ef2e : SUCCESS in 6m 40s\n- infra-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b69856a47a3b4bdfbe00892057b15ef2 : SUCCESS in 16m 59s\n- openstackci-beaker-puppet-4 https://zuul.opendev.org/t/openstack/build/896a688fb02540739f7bdc798fe8d6c5 : SUCCESS in 16m 27s\n- puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/f04e6add0e8c47759a18b4e278af85f6 : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1e5b59a0a51cc0b8eb40f74cefdd8e99840220c","revisions":{"85aaa6672ecfb80779d7ede5a896504dfcaee266":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:19:42.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/25/709525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/25/709525/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/1"}}},"commit":{"parents":[{"commit":"fb0da02f2c0f25c4665553472a17dbafe1a42ea8","subject":"Updated deployment scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/fb0da02f2c0f25c4665553472a17dbafe1a42ea8"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-02-24 13:19:06.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-02-24 13:19:06.000000000","tz":-180},"subject":"Removed stale config variables","message":"Removed stale config variables\n\nremoved all config variables related to\nss integration\n\nChange-Id: Id57bbf6329a0aac9904fd317ce0e1d8628993aa3\nSigned-off-by: smarcet \u003csmarcet@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/85aaa6672ecfb80779d7ede5a896504dfcaee266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/85aaa6672ecfb80779d7ede5a896504dfcaee266"}]},"branch":"refs/heads/master"},"b1e5b59a0a51cc0b8eb40f74cefdd8e99840220c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-24 18:31:53.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/25/709525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/25/709525/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/25/709525/2"}}},"commit":{"parents":[{"commit":"fb0da02f2c0f25c4665553472a17dbafe1a42ea8","subject":"Updated deployment scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/fb0da02f2c0f25c4665553472a17dbafe1a42ea8"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-02-24 13:19:06.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2020-02-24 18:31:34.000000000","tz":-180},"subject":"Removed stale config variables","message":"Removed stale config variables\n\nremoved all config variables related to\nss integration\n\nChange-Id: Id57bbf6329a0aac9904fd317ce0e1d8628993aa3\nSigned-off-by: smarcet \u003csmarcet@gmail.com\u003e\nDepends-on: https://review.opendev.org/#/c/709529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/b1e5b59a0a51cc0b8eb40f74cefdd8e99840220c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/b1e5b59a0a51cc0b8eb40f74cefdd8e99840220c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
