)]}'
{"id":"opendev%2Fpuppet-openstackid~772705","triplet_id":"opendev%2Fpuppet-openstackid~master~I0266509ad07d0d28c509b446f40cabc7ca7ab8ef","project":"opendev/puppet-openstackid","branch":"master","topic":"hotfix/fix_pip_version","hashtags":[],"change_id":"I0266509ad07d0d28c509b446f40cabc7ca7ab8ef","subject":"Fixed PIP version","status":"MERGED","created":"2021-01-27 15:14:36.000000000","updated":"2021-01-27 16:24:52.000000000","submitted":"2021-01-27 16:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772705-hotfix/fix_pip_version","meta_rev_id":"8f9034849e5b63e054f917bf33b9620761cc8413","_number":772705,"virtual_id_number":772705,"owner":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 16:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-27 15:48:51.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-27 15:48:51.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 15:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 15:46:58.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-01-27 15:48:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4c7ed78467106f8c7b6efb0f9665db2740ac8db5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2021-01-27 15:14:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47eab5e658e5a34fc0a7027a8f3716a6168263a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:33:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/c25d16028ab94a4db34b69c4296eb41d : SUCCESS in 5m 15s\n- system-config-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/953e57dfce49414da4ecb2f7c46cfcaa : SUCCESS in 11m 03s\n- system-config-puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/93d8cf3ad2e043a3a5578fd8fc49c4a3 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"721455c1582fd7fa45a61c3d8460b7affeb62382","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-27 15:46:58.000000000","message":"Patch Set 1:\n\nShould only be broken with new pip if using Python older than 3.6... do you have a link to a log/error? I suppose if it\u0027s running on Ubuntu 16.04 LTS (Xenial) then that could explain it.","accounts_in_message":[],"_revision_number":1},{"id":"8376a5c9e37580d37fcf54eb333c0f59d29b2945","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-27 15:48:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOkay, I see, the production server is still running Xenial. I\u0027m okay with pinning pip here since we\u0027ll need to replace all the Puppet with Ansible and container builds anyway to upgrade past Xenial.","accounts_in_message":[],"_revision_number":1},{"id":"1cd2c5d447dd4c09f1bedc9bf25737732c7c86b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c72f71c79f640340d55f4837a7a4cc3ccbe273ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- legacy-puppet-lint https://zuul.opendev.org/t/openstack/build/1186967c23924eca9015b8bd2d5a771f : SUCCESS in 6m 08s\n- system-config-puppet-apply-4-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/69b37cecd19f481d9d56f2e7e60608ca : SUCCESS in 16m 19s\n- system-config-puppet-beaker-rspec-puppet-4-infra https://zuul.opendev.org/t/openstack/build/d64581b80bb54e439d926bd4605136f4 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fbae9c050253c8dcf6b3aba746ac7841d9d997b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 16:18:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f9034849e5b63e054f917bf33b9620761cc8413","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2021-01-27 16:24:52.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Should only be broken with new pip if using Python older than 3.6... do you have a link to a log/error? I suppose if it\u0027s running on Ubuntu 16.04 LTS (Xenial) then that could explain it.\n\nyes \nits breaks here\n\nCollecting pip\n  Using cached https://files.pythonhosted.org/packages/de/47/58b9f3e6f611dfd17fb8bd9ed3e6f93b7ee662fb85bdfee3565e8979ddf7/pip-21.0-py3-none-any.whl\nInstalling collected packages: pip\n  Found existing installation: pip 8.1.1\n    Uninstalling pip-8.1.1:\n      Successfully uninstalled pip-8.1.1\nSuccessfully installed pip-21.0\ninstalling setuptools\nTraceback (most recent call last):\n  File \"/srv/openstackid/slot1/app/Console/Commands/SpammerProcess/env/bin/pip\", line 7, in \u003cmodule\u003e\n    from pip._internal.cli.main import main\n  File \"/srv/openstackid/slot1/app/Console/Commands/SpammerProcess/env/lib/python3.5/site-packages/pip/_internal/cli/main.py\", line 60\n    sys.stderr.write(f\"ERROR: {exc}\")\n                                   ^\nSyntaxError: invalid syntax","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca788b2a8bb6c4a7320e41cc920b0181ec39baae","revisions":{"ca788b2a8bb6c4a7320e41cc920b0181ec39baae":{"kind":"REWORK","_number":1,"created":"2021-01-27 15:14:36.000000000","uploader":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"ref":"refs/changes/05/772705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/puppet-openstackid","ref":"refs/changes/05/772705/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/05/772705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/05/772705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/puppet-openstackid refs/changes/05/772705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/puppet-openstackid refs/changes/05/772705/1"}}},"commit":{"parents":[{"commit":"ad9fa60259ccf9cbb5d881642d8f67144538899f","subject":"Changed default queue driver from database to redis to prevent deadlocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/ad9fa60259ccf9cbb5d881642d8f67144538899f"}]}],"author":{"name":"smarcet","email":"smarcet@gmail.com","date":"2021-01-27 15:13:54.000000000","tz":-180},"committer":{"name":"smarcet","email":"smarcet@gmail.com","date":"2021-01-27 15:13:54.000000000","tz":-180},"subject":"Fixed PIP version","message":"Fixed PIP version\n\nfixed pip install to 20.3.4 bc latest one\n(21.0.0) breaks python build\n\nChange-Id: I0266509ad07d0d28c509b446f40cabc7ca7ab8ef\nSigned-off-by: smarcet \u003csmarcet@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/ca788b2a8bb6c4a7320e41cc920b0181ec39baae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/puppet-openstackid/commit/ca788b2a8bb6c4a7320e41cc920b0181ec39baae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
