)]}'
{"id":"opendev%2Fpython-storyboardclient~574347","triplet_id":"opendev%2Fpython-storyboardclient~master~I89479c30c4118e46aed51f0f52630b704546892c","project":"opendev/python-storyboardclient","branch":"master","topic":"correct-distname","hashtags":[],"change_id":"I89479c30c4118e46aed51f0f52630b704546892c","subject":"fix tox python3 overrides","status":"NEW","created":"2018-06-11 17:09:02.000000000","updated":"2019-08-13 14:47:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9749d54f9330e194a32f9e673dd2d8f823dfc539","_number":574347,"virtual_id_number":574347,"owner":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},{"date":"2019-08-13 14:37:39.000000000","_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":1,"date":"2019-07-19 18:55:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 14:25:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-23 17:12:25.000000000","updated_by":{"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},"reviewer":{"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},"state":"REVIEWER"},{"updated":"2019-02-14 11:42:13.000000000","updated_by":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"reviewer":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"state":"REVIEWER"},{"updated":"2019-06-28 16:42:45.000000000","updated_by":{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},"reviewer":{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},"state":"REVIEWER"},{"updated":"2019-07-13 09:20:05.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 13:43:20.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-19 17:17:45.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-07-19 18:55:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 14:25:27.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-08-13 14:37:39.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"}],"messages":[{"id":"9f68482f36b4e58fdce2fbfc3342f2e5d8917787","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2018-06-11 17:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e03e32c9aabff490d04835ee5be1e57449332116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-11 17:51:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 http://logs.openstack.org/47/574347/1/check/tox-pep8/1bcf613/ : SUCCESS in 4m 32s\n- tox-py27 http://logs.openstack.org/47/574347/1/check/tox-py27/804548f/ : SUCCESS in 5m 06s\n- tox-py35 http://logs.openstack.org/47/574347/1/check/tox-py35/8d13d0e/ : SUCCESS in 5m 10s\n- build-sphinx-docs http://logs.openstack.org/47/574347/1/check/build-sphinx-docs/c337c66/html/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d5646dd0727aa324a8d73e35a3da794e34f198ae","author":{"_account_id":6088,"name":"Emmet Hikory","email":"persia@shipstone.jp","username":"persia"},"date":"2018-09-23 17:12:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2c7d9e471e700fea832c4607010fbcef08ae9e9","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-09-26 01:48:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"48b278395c54c9541c691254bd8ff41170f0fe6d","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2018-11-26 07:11:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"75c9bb53332cb57549d6c90985b8c7fc456f0dc7","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2018-11-26 07:11:57.000000000","message":"Topic set to bug/1801657","accounts_in_message":[],"_revision_number":2},{"id":"b606dcda7dd93a0cbc6ebe57797d4ddf9a42099e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 07:21:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 http://logs.openstack.org/47/574347/2/check/tox-pep8/bbb30c7/ : SUCCESS in 9m 25s\n- tox-py27 http://logs.openstack.org/47/574347/2/check/tox-py27/2ab98a6/ : SUCCESS in 3m 50s\n- tox-py35 http://logs.openstack.org/47/574347/2/check/tox-py35/589cd21/ : SUCCESS in 7m 47s\n- tox-docs http://logs.openstack.org/47/574347/2/check/tox-docs/2b4446f/html/ : SUCCESS in 4m 18s\n- tox-cover http://logs.openstack.org/47/574347/2/check/tox-cover/e26ca61/ : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":2},{"id":"974bdcee45be7600f0b4c6fa5ea5457ae2fa2d98","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-01-14 03:09:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ba2cfe57fe01e0e4493e5e6ff658ff33fe04a601","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-01-14 09:41:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"529f275f8c030b1de153e3f82daac1e0d81c9ed1","author":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"date":"2019-02-14 11:42:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d14467125e95890afbe30042726625cd525ca67c","author":{"_account_id":16960,"name":"Adam Coldrick","email":"adam.coldrick@codethink.co.uk","username":"SotK"},"date":"2019-06-28 16:42:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f2d1287ef54029df07dbaba0059d7171d90ef6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 16:42:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af6010a9c1c6069d6e7f5ae9c81ca1dae8f826ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 16:48:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/2/gate/tox-pep8/a159b3f/ : FAILURE in 3m 01s\n- tox-py27 http://logs.openstack.org/47/574347/2/gate/tox-py27/9d44261/ : SUCCESS in 3m 24s\n- tox-py35 http://logs.openstack.org/47/574347/2/gate/tox-py35/a146df8/ : SUCCESS in 3m 13s\n- tox-docs http://logs.openstack.org/47/574347/2/gate/tox-docs/883047a/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"94de521fc30e14c8f2f956df6a5bd72db09eff92","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-07-11 08:10:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bad43f066fd92764eceb2939adb3abb6c2e37150","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 08:18:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/2/check/tox-pep8/2fc7420/ : FAILURE in 3m 07s\n- tox-py27 http://logs.openstack.org/47/574347/2/check/tox-py27/2bdba35/ : SUCCESS in 2m 29s\n- tox-py35 http://logs.openstack.org/47/574347/2/check/tox-py35/6f3d702/ : SUCCESS in 3m 22s\n- tox-docs http://logs.openstack.org/47/574347/2/check/tox-docs/0883cb3/html/ : SUCCESS in 2m 19s\n- tox-cover http://logs.openstack.org/47/574347/2/check/tox-cover/467e3ee/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4d7327351716c08b07442aa5548e8847f063aeb5","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-11 08:56:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bef912d7e7a90c42981e5d75ed44af0a9dbc2693","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:04:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/2/check/tox-pep8/cdb115a/ : FAILURE in 3m 04s\n- tox-py27 http://logs.openstack.org/47/574347/2/check/tox-py27/498b24a/ : SUCCESS in 2m 42s\n- tox-py35 http://logs.openstack.org/47/574347/2/check/tox-py35/c8640f5/ : SUCCESS in 3m 12s\n- tox-docs http://logs.openstack.org/47/574347/2/check/tox-docs/4c97fba/html/ : SUCCESS in 2m 29s\n- tox-cover http://logs.openstack.org/47/574347/2/check/tox-cover/c80caee/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ac855adf2827033304b0af933aec08d6f0b74532","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-12 01:45:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f7943bd519a5ee29680656bd6a36d653538f407f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:52:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/2/check/tox-pep8/2f45dee/ : FAILURE in 2m 22s\n- tox-py27 http://logs.openstack.org/47/574347/2/check/tox-py27/fa12ddd/ : SUCCESS in 2m 25s\n- tox-py35 http://logs.openstack.org/47/574347/2/check/tox-py35/d8e3188/ : SUCCESS in 2m 44s\n- tox-docs http://logs.openstack.org/47/574347/2/check/tox-docs/fdac872/html/ : SUCCESS in 3m 11s\n- tox-cover http://logs.openstack.org/47/574347/2/check/tox-cover/a7c9772/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b540fd9c991bdb58c5710483d2fd37c473a76fe3","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-07-12 01:57:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"545c6abf50be5130577c0bf74c53a4446be48879","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 02:04:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/3/check/tox-pep8/48e52cc/ : FAILURE in 2m 20s\n- tox-py27 http://logs.openstack.org/47/574347/3/check/tox-py27/4d80c23/ : SUCCESS in 2m 20s\n- tox-py35 http://logs.openstack.org/47/574347/3/check/tox-py35/95e7424/ : SUCCESS in 3m 32s\n- tox-docs http://logs.openstack.org/47/574347/3/check/tox-docs/c94926d/html/ : SUCCESS in 1m 54s\n- tox-cover http://logs.openstack.org/47/574347/3/check/tox-cover/dbaf152/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"532380b3ac3c86b5f95f0eb9a59350f281051957","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-07-13 09:17:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c4fb92f4f0ba573924c529ede3da1d8efe43153c","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-13 09:20:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2745c3703f3aac4440874646c9ce06504fedb8f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 09:25:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/3/check/tox-pep8/900a041/ : FAILURE in 2m 42s\n- tox-py27 http://logs.openstack.org/47/574347/3/check/tox-py27/26ea1dc/ : SUCCESS in 2m 20s\n- tox-py35 http://logs.openstack.org/47/574347/3/check/tox-py35/ef33a8a/ : SUCCESS in 2m 21s\n- tox-docs http://logs.openstack.org/47/574347/3/check/tox-docs/a4a8578/html/ : SUCCESS in 2m 31s\n- tox-cover http://logs.openstack.org/47/574347/3/check/tox-cover/eaa443d/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e820231e613afd88de3c06bc62db4919da9c04ac","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-13 11:48:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"917a71f4e2b3a97bcbbca1d86ae823667f9144ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 11:56:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/3/check/tox-pep8/8e5d50d/ : FAILURE in 2m 36s\n- tox-py27 http://logs.openstack.org/47/574347/3/check/tox-py27/b6394be/ : SUCCESS in 2m 16s\n- tox-py35 http://logs.openstack.org/47/574347/3/check/tox-py35/7b839af/ : SUCCESS in 3m 17s\n- tox-docs http://logs.openstack.org/47/574347/3/check/tox-docs/224c239/html/ : SUCCESS in 1m 58s\n- tox-cover http://logs.openstack.org/47/574347/3/check/tox-cover/656263a/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"834719ccf5757f828adf2c7c3ec96fbfeee8420b","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 08:12:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3ce3a503bfde6736f415e1ba9397c857998fb014","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 08:16:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/3/check/tox-pep8/77780aa/ : FAILURE in 2m 32s\n- tox-py27 http://logs.openstack.org/47/574347/3/check/tox-py27/49cb834/ : SUCCESS in 2m 42s\n- tox-py35 http://logs.openstack.org/47/574347/3/check/tox-py35/87e90c8/ : SUCCESS in 3m 04s\n- tox-docs http://logs.openstack.org/47/574347/3/check/tox-docs/84fb1a2/html/ : SUCCESS in 2m 27s\n- tox-cover http://logs.openstack.org/47/574347/3/check/tox-cover/4bc0570/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b244c3028264c8ac07824a9d39020df72b195d73","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 12:25:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6d3276bc1e9b34b4bef212df697bcc63fa2ac5d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 12:32:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/3/check/tox-pep8/52f43f0/ : FAILURE in 2m 36s\n- tox-py27 http://logs.openstack.org/47/574347/3/check/tox-py27/c4c1c75/ : SUCCESS in 2m 31s\n- tox-py35 http://logs.openstack.org/47/574347/3/check/tox-py35/195268a/ : SUCCESS in 3m 31s\n- tox-docs http://logs.openstack.org/47/574347/3/check/tox-docs/9af2254/html/ : SUCCESS in 2m 03s\n- tox-cover http://logs.openstack.org/47/574347/3/check/tox-cover/95a28c8/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ed0e56ed4167e403d5b85c89c0ba628dcbd62b6c","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-14 13:43:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"936a5d1b90106b8efeaff2d4e68a8178ce17682b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 13:49:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/3/check/tox-pep8/a5c9de0/ : FAILURE in 3m 12s\n- tox-py27 http://logs.openstack.org/47/574347/3/check/tox-py27/6675444/ : SUCCESS in 3m 49s\n- tox-py35 http://logs.openstack.org/47/574347/3/check/tox-py35/3775109/ : SUCCESS in 4m 10s\n- tox-docs http://logs.openstack.org/47/574347/3/check/tox-docs/8883f70/html/ : SUCCESS in 3m 01s\n- tox-cover http://logs.openstack.org/47/574347/3/check/tox-cover/1f3724f/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"d5572d43eb642b7ce91098d56146d23b175ad647","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-19 17:17:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9a33d84745e487e368bfa57a794e9c4a8858dabd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:37:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-pep8 http://logs.openstack.org/47/574347/4/check/tox-pep8/8ab7727/ : FAILURE in 4m 27s\n- tox-py27 http://logs.openstack.org/47/574347/4/check/tox-py27/e4bc5dc/ : SUCCESS in 4m 07s\n- tox-py35 http://logs.openstack.org/47/574347/4/check/tox-py35/354ed18/ : SUCCESS in 3m 49s\n- tox-docs http://logs.openstack.org/47/574347/4/check/tox-docs/aa9a6a2/html/ : SUCCESS in 2m 33s\n- tox-cover http://logs.openstack.org/47/574347/4/check/tox-cover/76aa476/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5dd2808d8a050774126902c20978873f987cb4b5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-19 18:41:24.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"60d753663ee031be553d04085cdbd9b5f1cd1083","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:55:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 http://logs.openstack.org/47/574347/5/check/tox-pep8/9dc2b97/ : SUCCESS in 4m 02s\n- tox-py27 http://logs.openstack.org/47/574347/5/check/tox-py27/98ab619/ : SUCCESS in 3m 22s\n- tox-py35 http://logs.openstack.org/47/574347/5/check/tox-py35/750ed1e/ : SUCCESS in 3m 02s\n- tox-docs http://logs.openstack.org/47/574347/5/check/tox-docs/8537b14/html/ : SUCCESS in 2m 33s\n- tox-cover http://logs.openstack.org/47/574347/5/check/tox-cover/e185c14/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":5},{"id":"7ed98ebb93fcc3ca6751c45fa99193029e8d5860","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-08-13 14:25:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"404c8037e814113a798980ba4a5c74eecfc78cf5","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-08-13 14:37:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"658b73d7af8faddf1366defbc10eb6363ddc1d9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:47:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/959f388479fb41c18b447bd0edc28733 : SUCCESS in 4m 55s\n- tox-py27 https://zuul.opendev.org/t/openstack/build/f353de8c7e3e46179ffd3f872c6b4c1d : SUCCESS in 4m 56s\n- tox-py35 https://zuul.opendev.org/t/openstack/build/bc0fb5b5f9be4245a200556c9c3312a5 : SUCCESS in 3m 04s\n- tox-docs https://zuul.opendev.org/t/openstack/build/5cbf954622ac4b5885164e97eb6866ef : SUCCESS in 2m 19s\n- tox-cover https://zuul.opendev.org/t/openstack/build/9a641d98472748fea79b01e00ef66713 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0be30defd0981172bf6add8bbc68c028f6029f32","revisions":{"827c50273615e7fc810c5571570c580d18746e5f":{"kind":"REWORK","_number":1,"created":"2018-06-11 17:09:02.000000000","uploader":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"ref":"refs/changes/47/574347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/python-storyboardclient","ref":"refs/changes/47/574347/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/1"}}},"commit":{"parents":[{"commit":"af25ae6e07531422abee43150a535b4176bb9caa","subject":"remove sphinx-isms from README.rst and clean up docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/af25ae6e07531422abee43150a535b4176bb9caa"}]}],"author":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-06-11 17:08:08.000000000","tz":480},"committer":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-06-11 17:08:53.000000000","tz":480},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I89479c30c4118e46aed51f0f52630b704546892c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/827c50273615e7fc810c5571570c580d18746e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/827c50273615e7fc810c5571570c580d18746e5f"}]},"branch":"refs/heads/master"},"ee89b2bfd08b1750ec196ecb260c9506b7c28385":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-26 07:11:48.000000000","uploader":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"ref":"refs/changes/47/574347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/python-storyboardclient","ref":"refs/changes/47/574347/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/2"}}},"commit":{"parents":[{"commit":"af25ae6e07531422abee43150a535b4176bb9caa","subject":"remove sphinx-isms from README.rst and clean up docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/af25ae6e07531422abee43150a535b4176bb9caa"}]}],"author":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-06-11 17:08:08.000000000","tz":480},"committer":{"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","date":"2018-11-26 07:11:46.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I89479c30c4118e46aed51f0f52630b704546892c\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/ee89b2bfd08b1750ec196ecb260c9506b7c28385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/ee89b2bfd08b1750ec196ecb260c9506b7c28385"}]},"branch":"refs/heads/master"},"5107d936c6fea890039a2ec07fe196da90ef0446":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-12 01:57:11.000000000","uploader":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"ref":"refs/changes/47/574347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/python-storyboardclient","ref":"refs/changes/47/574347/3","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/3"}}},"commit":{"parents":[{"commit":"08e5be6ff69ca3bce794dff0ed65e64cc532b9c0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/08e5be6ff69ca3bce794dff0ed65e64cc532b9c0"}]}],"author":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-06-11 17:08:08.000000000","tz":480},"committer":{"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","date":"2019-07-12 01:57:11.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: I89479c30c4118e46aed51f0f52630b704546892c\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/5107d936c6fea890039a2ec07fe196da90ef0446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/5107d936c6fea890039a2ec07fe196da90ef0446"}]},"branch":"refs/heads/master"},"714487fae67fe9550bed741485da16ce02c6ae96":{"kind":"REWORK","_number":4,"created":"2019-07-19 17:17:45.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/47/574347/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/python-storyboardclient","ref":"refs/changes/47/574347/4","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/4"}}},"commit":{"parents":[{"commit":"08e5be6ff69ca3bce794dff0ed65e64cc532b9c0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/08e5be6ff69ca3bce794dff0ed65e64cc532b9c0"}]}],"author":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-06-11 17:08:08.000000000","tz":480},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-07-19 17:01:15.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in the meta testenv.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nAdd a new option from tox 3.1 to ignore this basepython override\nconflict, and update the minver accordingly.\n\nChange-Id: I89479c30c4118e46aed51f0f52630b704546892c\nTask: #23643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/714487fae67fe9550bed741485da16ce02c6ae96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/714487fae67fe9550bed741485da16ce02c6ae96"}]},"branch":"refs/heads/master"},"0be30defd0981172bf6add8bbc68c028f6029f32":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-19 18:41:24.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/47/574347/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/python-storyboardclient","ref":"refs/changes/47/574347/5","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/python-storyboardclient refs/changes/47/574347/5"}}},"commit":{"parents":[{"commit":"4fa187110d45f8197032d835d1e83e64de375b57","subject":"Correct the distname in PBR version discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/4fa187110d45f8197032d835d1e83e64de375b57"}]}],"author":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-06-11 17:08:08.000000000","tz":480},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-07-19 18:41:12.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in the meta testenv.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nAdd a new option from tox 3.1 to ignore this basepython override\nconflict, and update the minver accordingly.\n\nChange-Id: I89479c30c4118e46aed51f0f52630b704546892c\nTask: #23643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/0be30defd0981172bf6add8bbc68c028f6029f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/python-storyboardclient/commit/0be30defd0981172bf6add8bbc68c028f6029f32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
