)]}'
{"id":"opendev%2Fsandbox~824708","triplet_id":"opendev%2Fsandbox~master~I7cbb745bb673d20412240adf1e170325c76be12d","project":"opendev/sandbox","branch":"master","topic":"apavlov","attention_set":{"34465":{"account":{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},"last_update":"2022-01-14 13:31:22.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7cbb745bb673d20412240adf1e170325c76be12d","subject":"apavlov test","status":"NEW","created":"2022-01-14 13:29:56.000000000","updated":"2022-01-14 13:31:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6b0453b004dae40c58902408023e84eb9de4017","_number":824708,"virtual_id_number":824708,"owner":{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-14 13:31:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 13:31:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 13:31:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"b51a398f299f46fc81cf4ec0169a0278149605af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},"date":"2022-01-14 13:29:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfc1066af0f97d559e22f14ee302b81532d8708c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 13:31:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/7b6ac0c734914cedb74b01bbb239d7ef : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"c6b0453b004dae40c58902408023e84eb9de4017","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-14 13:31:22.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50dd98e35893dd7eea9816e3afb745fbea133668","revisions":{"50dd98e35893dd7eea9816e3afb745fbea133668":{"kind":"REWORK","_number":1,"created":"2022-01-14 13:29:56.000000000","uploader":{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},"ref":"refs/changes/08/824708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/sandbox","ref":"refs/changes/08/824708/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/sandbox refs/changes/08/824708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/sandbox refs/changes/08/824708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/sandbox refs/changes/08/824708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/sandbox refs/changes/08/824708/1"}}},"commit":{"parents":[{"commit":"560e5db47499fbe60b9dfcc30f4596c177ff83d2","subject":"hello_from_lesson.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/sandbox/commit/560e5db47499fbe60b9dfcc30f4596c177ff83d2"}]}],"author":{"name":"a.pavlov","email":"a.pavlov@yadro.com","date":"2022-01-14 11:57:02.000000000","tz":180},"committer":{"name":"a.pavlov","email":"a.pavlov@yadro.com","date":"2022-01-14 13:29:53.000000000","tz":180},"subject":"apavlov test","message":"apavlov test\n\nChange-Id: I7cbb745bb673d20412240adf1e170325c76be12d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/sandbox/commit/50dd98e35893dd7eea9816e3afb745fbea133668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/sandbox/commit/50dd98e35893dd7eea9816e3afb745fbea133668"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
