)]}'
{"id":"opendev%2Fsandbox~898708","triplet_id":"opendev%2Fsandbox~master~I229d8841108e9840d7e4d203ee1ce4e3f18761bf","project":"opendev/sandbox","branch":"master","topic":"new-branch","hashtags":[],"change_id":"I229d8841108e9840d7e4d203ee1ce4e3f18761bf","subject":"Test-file","status":"ABANDONED","created":"2023-10-18 06:37:08.000000000","updated":"2023-10-30 05:53:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"14279aed75f31887bc79de929cfdc4a8a7dc3800","_number":898708,"virtual_id_number":898708,"owner":{"_account_id":36036,"name":"ipatini","display_name":"ipatini","email":"ipatini@mail.ntua.gr","username":"ipatini"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-18 06:37:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-18 06:37:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2d343ea7c5a32373dcdbd20e5f0d6a2b94f06bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36036,"name":"ipatini","display_name":"ipatini","email":"ipatini@mail.ntua.gr","username":"ipatini"},"date":"2023-10-18 06:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b2f63f2da8742a2d5578bd1521426f793e8a166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 06:37:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57d1ba673714e6db0706df31ed687dc\n\n- noop https://zuul.opendev.org/t/openstack/build/424f8ad4bd2941d4857e0f3efe54673f : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"14279aed75f31887bc79de929cfdc4a8a7dc3800","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36036,"name":"ipatini","display_name":"ipatini","email":"ipatini@mail.ntua.gr","username":"ipatini"},"date":"2023-10-30 05:53:48.000000000","message":"Abandoned\n\ntest","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69c8e358816579470db4ee266ad34794d61e6743","revisions":{"69c8e358816579470db4ee266ad34794d61e6743":{"kind":"REWORK","_number":1,"created":"2023-10-18 06:37:08.000000000","uploader":{"_account_id":36036,"name":"ipatini","display_name":"ipatini","email":"ipatini@mail.ntua.gr","username":"ipatini"},"ref":"refs/changes/08/898708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/sandbox","ref":"refs/changes/08/898708/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/sandbox refs/changes/08/898708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/sandbox refs/changes/08/898708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/sandbox refs/changes/08/898708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/sandbox refs/changes/08/898708/1"}}},"commit":{"parents":[{"commit":"b46cf866db31a3621341c89b24481c644f854fad","subject":"Testing commit to show gerrit worklow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/sandbox/commit/b46cf866db31a3621341c89b24481c644f854fad"}]}],"author":{"name":"ipatini","email":"ipatini@mail.ntua.gr","date":"2023-10-18 06:36:29.000000000","tz":180},"committer":{"name":"ipatini","email":"ipatini@mail.ntua.gr","date":"2023-10-18 06:36:29.000000000","tz":180},"subject":"Test-file","message":"Test-file\n\nChange-Id: I229d8841108e9840d7e4d203ee1ce4e3f18761bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/sandbox/commit/69c8e358816579470db4ee266ad34794d61e6743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/sandbox/commit/69c8e358816579470db4ee266ad34794d61e6743"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
