)]}'
{"id":"opendev%2Fsubunit2sql~764832","triplet_id":"opendev%2Fsubunit2sql~master~I219755d98d9f8b5a284c0e858b60f852b948c8d4","project":"opendev/subunit2sql","branch":"master","topic":"fix-master","hashtags":[],"change_id":"I219755d98d9f8b5a284c0e858b60f852b948c8d4","subject":"Fix compatibility with latest oslo.config","status":"MERGED","created":"2020-11-30 22:19:10.000000000","updated":"2021-01-22 20:21:55.000000000","submitted":"2021-01-22 20:19:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"764832-fix-master","meta_rev_id":"4c2fd2b1b884bde16348dec7a9b1313ebf0b5b72","_number":764832,"virtual_id_number":764832,"owner":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 20:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-01-22 15:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-22 20:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-22 20:09:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 22:19:59.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-11-30 22:21:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-11-30 22:23:37.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"CC"},{"updated":"2020-11-30 22:30:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-30 22:41:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 16:01:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-01-21 20:30:29.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-01-22 15:01:21.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"}],"messages":[{"id":"c1755b593379f8c4d6b3c8bc4e23b4821cdd1c6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-11-30 22:19:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e9b87b543070365694cfabacbbe76c372d4b03f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-11-30 22:21:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7aa672ae4134aa203e6ca5e6f717e09ce7eb6662","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-11-30 22:23:37.000000000","message":"Assignee added: Masayuki Igawa \u003cmasayuki@igawa.io\u003e","accounts_in_message":[],"_revision_number":1},{"id":"c1d9b1a165d4ff2cf9a7cd707c77343aff20b1a0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-30 22:30:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0379168c30f8d0fc977c4aec3c0565db511d7425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 22:41:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c95fd3f00cf4a10b8047a1c6aa41db9 : RETRY_LIMIT in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322526630dea43d3ac479d10428a8464 : FAILURE in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1335cd7159c0417f9023d073f1f99dc8 : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ec7829144574f56835f334fc35ca36f : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92975d78acd44efb905f73f94f4969e6 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd541c3f741349f88c66b7d80dbf657f : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e8caf7396450a5998224576adb0b0b42644a8d38","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-11-30 22:45:12.000000000","message":"Patch Set 1:\n\nLooks like we may have some bitrot to untangle in the jobs for this repo.","accounts_in_message":[],"_revision_number":1},{"id":"65a336ac55d7da010dfcf13fd29ec6ddadf3c28d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-12-01 17:14:01.000000000","message":"Patch Set 1: Code-Review+1\n\nAccording to the release note, this was done to match argparse\u0027s behavior: https://docs.openstack.org/releasenotes/oslo.config/ussuri.html#relnotes-7-0-0-stable-ussuri-upgrade-notes","accounts_in_message":[],"_revision_number":1},{"id":"254f60d786b6ca831a99debd7525b7e99d64f3e5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-21 16:01:35.000000000","message":"Patch Set 1:\n\nrecheck getting new results","accounts_in_message":[],"_revision_number":1},{"id":"fb26f5706d9e383c015a78586103c926282a010a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 16:31:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4e077995fba4247aa63a569cb883b41 : RETRY_LIMIT in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d3e89d75194509bc6aae944c2c62d2 : FAILURE in 14m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79e16885b82c40338ec694ce8bd1ad2e : FAILURE in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1518179d1c424358a5a299d588eb3ad2 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9176ee31c3a24fd2ba1d70ae15399026 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7e4e74e9ce4424aef06ed5c151da3e : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"512dacc6c4a169b7a682090a950b3418df3976fe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-21 20:29:19.000000000","message":"Patch Set 1:\n\nfailing jobs error due to py3 compatibility which need to be squashed in this fox to pass gate all together. \n\nopenstack-tox-cover: with mysql 8.0 there is no implicit user creation with GRANT.we need to create user fist.\n\nopenstack-tox-pep8 hacking version used is too old which pull old pyflakes (\u003c 2.1.0) not compatible with py3.8\n\nopenstack-tox-py27 this need to cap stestr with \u003c3.0.0 for py2 support.","accounts_in_message":[],"_revision_number":1},{"id":"d47b94494de19a78adcbd881ff9766308d942ae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-21 20:30:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d0225b791e8acb2414740f284ae3cee441f11a9f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-21 20:43:38.000000000","message":"Patch Set 2:\n\nseems one more to fix \n\n ERROR: Requested docutils\u003e\u003d0.12 from https://mirror-int.iad.rax.opendev.org/pypifiles/packages/d4/12/6c3fd74a590c7327c98cae008c11d536029fa9cd7924de477e8cb8804186/docutils-0.15.1-post1.tar.gz#sha256\u003df33ddb723332c6d6b6d99731ee1fc0c35eb4044a2df5cca1c64c8aa78eaf22cb (from sphinx!\u003d1.6.6,!\u003d1.6.7,!\u003d2.1.0-\u003e-r /home/zuul/src/opendev.org/opendev/subunit2sql/test-requirements.txt (line 10)) has different version in metadata: \u00270.15.1\u0027","accounts_in_message":[],"_revision_number":2},{"id":"648709c2e333b420832cf0fba1802a6cc3d6d771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:57:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f42f31cb2f045279b22ad7a73db9c8e : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdd27d0c2d714fbfb11516a07eca8621 : FAILURE in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76c72542a82344c980846361b2650d8e : SUCCESS in 22m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af8fc985c3a94e11a883f87799155376 : FAILURE in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec066a07a56f4180a1b5a813d88dee53 : FAILURE in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0001869d4f094c68866dd0076e08e8ca : RETRY_LIMIT in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc99a4eaa38e425e84075d5804a3a719 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4f3a6492cb6599ff13eeff1e56fcbc7cc147eeff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-22 00:41:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee2425c2a9365ab6f27f975c60de4c80b789149e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-22 01:30:52.000000000","message":"Patch Set 3:\n\nseems like we have to bump the hacking to \u003e2.0.0 otherwise pip is not able to resolve the deps.","accounts_in_message":[],"_revision_number":3},{"id":"5204a4e274b778be99086374ed343050d5358d31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-22 01:45:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96fff349c87ceac50292b7b32b6251c75ea058e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 02:23:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97246c114f67460bb002ac18dcd95c34 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51dada522f2b4b358d0a0a0e085011d6 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65378f7b215b4e609362362b9620c272 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d73710b0a2546a09942dce018322438 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29e163c77eaa4469bd665d5ff93ffc61 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3f338b0b2f44e12aaddcf7a0d0972e3 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785cc339208b4f7d96b9307fe5a3fb79 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":4},{"id":"7fe7576e9d43798ba147bb63ef8767a63100dcdc","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2021-01-22 15:01:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nThanks. lgtm","accounts_in_message":[],"_revision_number":4},{"id":"c58c0c0839edb22b4d1b8e8bf40f8f7c8bf048c4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-22 20:09:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fcd19bd995c3a68188f7a6f5e926eb29c161567d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:09:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0ef808e8e5273849a80c088cbcf87aff3b33a6b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:19:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91321198716241e68645b7934e60955f : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17b2b15519674386ba8c8e74c969f1ed : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97c2c63fcbf84eabbe3ef27f8ae6c341 : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ace2bdf82e2c4b13a8e549c530e8d017 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb49ed7dc724482bd196c486f5129fe : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcee32961e14872a85fef07003a0365 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c245e8d3d1a96cc0ece6a0161bb5c22e0af31c71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:19:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4c2fd2b1b884bde16348dec7a9b1313ebf0b5b72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:21:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffbddcdc3c5c488bbb517eb55ae6b332 : SUCCESS in 1m 35s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b731beb47ca448aa9189765fadc6ab5 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fca99f945cffe53472d62e5695df352c6d40e5e6","revisions":{"fdbad748644ada78736aaa2afc47c23f66ad315a":{"kind":"REWORK","_number":1,"created":"2020-11-30 22:19:10.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/32/764832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/subunit2sql","ref":"refs/changes/32/764832/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/1"}}},"commit":{"parents":[{"commit":"99f83fda8742c1c4732c832d139f1c68dce36da9","subject":"Merge \"Add .zuul.yaml and python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/99f83fda8742c1c4732c832d139f1c68dce36da9"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-11-30 22:14:13.000000000","tz":-300},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-11-30 22:16:49.000000000","tz":-300},"subject":"Fix compatibility with latest oslo.config","message":"Fix compatibility with latest oslo.config\n\nWith recent versions of oslo.config the tests have started to fail\nbecause the multistring opt used for specifying input subunit file paths\nwas getting an implicit required. However it\u0027s not a required field just\npositional. This commit fixes this issue by explicitly setting it to not\nbe requried and setting a default value of None for it. I\u0027m not sure when\nan implicit required was added in oslo.config, but this should fix it\nregardless of which oslo.config version is used.\n\nStory: 2008348\nTask: 41250\n\nChange-Id: I219755d98d9f8b5a284c0e858b60f852b948c8d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/fdbad748644ada78736aaa2afc47c23f66ad315a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/fdbad748644ada78736aaa2afc47c23f66ad315a"}]},"branch":"refs/heads/master"},"2853bc7806b74accc601ef0ba0c9cf5d8ba79dd1":{"kind":"REWORK","_number":2,"created":"2021-01-21 20:30:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/764832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/subunit2sql","ref":"refs/changes/32/764832/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/2"}}},"commit":{"parents":[{"commit":"99f83fda8742c1c4732c832d139f1c68dce36da9","subject":"Merge \"Add .zuul.yaml and python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/99f83fda8742c1c4732c832d139f1c68dce36da9"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-11-30 22:14:13.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-01-21 20:30:28.000000000","tz":-360},"subject":"Fix compatibility with latest oslo.config","message":"Fix compatibility with latest oslo.config\n\nWith recent versions of oslo.config the tests have started to fail\nbecause the multistring opt used for specifying input subunit file paths\nwas getting an implicit required. However it\u0027s not a required field just\npositional. This commit fixes this issue by explicitly setting it to not\nbe requried and setting a default value of None for it. I\u0027m not sure when\nan implicit required was added in oslo.config, but this should fix it\nregardless of which oslo.config version is used.\n\nFixing other gate failure also for pythin 2.7 \u0026 3.8 compatibility.\n\n- with mysql 8.0 there is no implicit user creation with GRANT. we\nneed to create user first.\n\n- hacking version used is too old which pull old pyflakes (\u003c 2.1.0) not\ncompatible with py3.8. Bumping hacking version to 1.1.0 (not latest which needs\nlot of updates in code) so that it allow to pull the 2.1.1 pyflakes.\n\nfor openstack-tox-py27 we need to cap stestr with \u003c3.0.0 so that it can pull\nstestr py2 compatible version.\n\nStory: 2008348\nTask: 41250\n\nChange-Id: I219755d98d9f8b5a284c0e858b60f852b948c8d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/2853bc7806b74accc601ef0ba0c9cf5d8ba79dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/2853bc7806b74accc601ef0ba0c9cf5d8ba79dd1"}]},"branch":"refs/heads/master"},"6163992b3dd9059072afa352c0e1a471b677c5e4":{"kind":"REWORK","_number":3,"created":"2021-01-22 00:41:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/764832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/subunit2sql","ref":"refs/changes/32/764832/3","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/3"}}},"commit":{"parents":[{"commit":"99f83fda8742c1c4732c832d139f1c68dce36da9","subject":"Merge \"Add .zuul.yaml and python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/99f83fda8742c1c4732c832d139f1c68dce36da9"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-11-30 22:14:13.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-01-22 00:40:12.000000000","tz":-360},"subject":"Fix compatibility with latest oslo.config","message":"Fix compatibility with latest oslo.config\n\nWith recent versions of oslo.config the tests have started to fail\nbecause the multistring opt used for specifying input subunit file paths\nwas getting an implicit required. However it\u0027s not a required field just\npositional. This commit fixes this issue by explicitly setting it to not\nbe requried and setting a default value of None for it. I\u0027m not sure when\nan implicit required was added in oslo.config, but this should fix it\nregardless of which oslo.config version is used.\n\nFixing other gate failure also for pythin 2.7 \u0026 3.8 compatibility.\n\n- with mysql 8.0 there is no implicit user creation with GRANT. we\nneed to create user first.\n\n- hacking version used is too old which pull old pyflakes (\u003c 2.1.0) not\ncompatible with py3.8. Bumping hacking version to 1.1.0 (not latest which needs\nlot of updates in code) so that it allow to pull the 2.1.1 pyflakes.\n\n- for openstack-tox-py27 we need to cap stestr with \u003c3.0.0 so that it can pull\nstestr py2 compatible version.\n\n- move dec requirements seprate from test-requirements otherwise it conflist with\nother deps.\n\nStory: 2008348\nTask: 41250\n\nChange-Id: I219755d98d9f8b5a284c0e858b60f852b948c8d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/6163992b3dd9059072afa352c0e1a471b677c5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/6163992b3dd9059072afa352c0e1a471b677c5e4"}]},"branch":"refs/heads/master"},"fca99f945cffe53472d62e5695df352c6d40e5e6":{"kind":"REWORK","_number":4,"created":"2021-01-22 01:45:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/764832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/subunit2sql","ref":"refs/changes/32/764832/4","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/subunit2sql refs/changes/32/764832/4"}}},"commit":{"parents":[{"commit":"99f83fda8742c1c4732c832d139f1c68dce36da9","subject":"Merge \"Add .zuul.yaml and python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/99f83fda8742c1c4732c832d139f1c68dce36da9"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-11-30 22:14:13.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-01-22 01:43:39.000000000","tz":-360},"subject":"Fix compatibility with latest oslo.config","message":"Fix compatibility with latest oslo.config\n\nWith recent versions of oslo.config the tests have started to fail\nbecause the multistring opt used for specifying input subunit file paths\nwas getting an implicit required. However it\u0027s not a required field just\npositional. This commit fixes this issue by explicitly setting it to not\nbe requried and setting a default value of None for it. I\u0027m not sure when\nan implicit required was added in oslo.config, but this should fix it\nregardless of which oslo.config version is used.\n\nFixing other gate failure also for pythin 2.7 \u0026 3.8 compatibility.\n\n- with mysql 8.0 there is no implicit user creation with GRANT. we\nneed to create user first.\n\n- hacking version used is too old which pull old pyflakes (\u003c 2.1.0) not\ncompatible with py3.8. Bumping hacking version to 2.0.0 which solve the\npyflakes version conflict with new pip resolver but skipping all the\nnew checks of 2.0.0. We can fix those later once gate is green.\n\n- for openstack-tox-py27 we need to cap stestr with \u003c3.0.0 so that it can pull\nstestr py2 compatible version.\n\n- move dec requirements seprate from test-requirements otherwise it conflist with\nother deps.\n\nStory: 2008348\nTask: 41250\n\nChange-Id: I219755d98d9f8b5a284c0e858b60f852b948c8d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/fca99f945cffe53472d62e5695df352c6d40e5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/subunit2sql/commit/fca99f945cffe53472d62e5695df352c6d40e5e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
