)]}'
{"id":"opendev%2Fsystem-config~967708","triplet_id":"opendev%2Fsystem-config~master~Ida9f977df33f4c52ca3542e83d8593242406cae0","project":"opendev/system-config","branch":"master","hashtags":[],"change_id":"Ida9f977df33f4c52ca3542e83d8593242406cae0","subject":"DNM: Trigger some channel log collisions","status":"ABANDONED","created":"2025-11-19 15:41:40.000000000","updated":"2025-12-05 19:46:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"995dacc5b66911d261b059fa759482387c394ebc","_number":967708,"virtual_id_number":967708,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-19 16:04:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-11-19 16:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e7c5239c832d2b4d61c0240d9fd0ff4ada7d179","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-11-19 15:41:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6bbd7f26cd29d0868d1ae76b17e830290890256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 16:04:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b2cf03262184df8b8fdef21461a496c\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ebf0c7d0dfa24701bddae2096d69a885 : SUCCESS in 2m 48s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/69bbdbc2c6f6437780406fe854fc3ecb : SUCCESS in 21m 59s\n- tox-linters https://zuul.opendev.org/t/openstack/build/d377804883604375b912906a6ae92d95 : FAILURE in 3m 58s\n- system-config-run-base https://zuul.opendev.org/t/openstack/build/48a7c6a5a738403792a2f8a16ec8784e : SUCCESS in 10m 20s\n- system-config-run-base-ansible-devel https://zuul.opendev.org/t/openstack/build/2b74c0ae006d4217a762babc09c3fc98 : POST_FAILURE in 4m 40s (non-voting)\n- system-config-run-eavesdrop https://zuul.opendev.org/t/openstack/build/96334ffb8e034466ab3e1fd94d468201 : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b421a685ec4161db5c2ee11d3a1af44e1528d1cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 16:04:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd45a5865dc5409fa13df2bef49a8e1e\n\n- system-config-run-base-arm64 https://zuul.opendev.org/t/openstack/build/2dcb76e6c0b14643a7fc491f99cb8c0b : SUCCESS in 20m 21s","accounts_in_message":[],"_revision_number":1},{"id":"995dacc5b66911d261b059fa759482387c394ebc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-12-05 19:46:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"671cc71716c1c5eeeaaa5d12b76c98623c2f4297","revisions":{"671cc71716c1c5eeeaaa5d12b76c98623c2f4297":{"kind":"REWORK","_number":1,"created":"2025-11-19 15:41:40.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/08/967708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/system-config","ref":"refs/changes/08/967708/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/system-config refs/changes/08/967708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/system-config refs/changes/08/967708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/system-config refs/changes/08/967708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/system-config refs/changes/08/967708/1"}}},"commit":{"parents":[{"commit":"e4bc82e790ded822919ce6c9bff6afaa7e6e3e90","subject":"Add checks to avoid irc and matrix log collisions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/system-config/commit/e4bc82e790ded822919ce6c9bff6afaa7e6e3e90"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-11-19 15:40:59.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-11-19 15:40:59.000000000","tz":0},"subject":"DNM: Trigger some channel log collisions","message":"DNM: Trigger some channel log collisions\n\nChange-Id: Ida9f977df33f4c52ca3542e83d8593242406cae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/system-config/commit/671cc71716c1c5eeeaaa5d12b76c98623c2f4297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/system-config/commit/671cc71716c1c5eeeaaa5d12b76c98623c2f4297"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
