)]}'
{"id":"opendev%2Fzone-gating.dev~880906","triplet_id":"opendev%2Fzone-gating.dev~master~I1ccec6900537c6c4243816dcd96b5fb9427f6bb5","project":"opendev/zone-gating.dev","branch":"master","topic":"dns-jammy","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-04-25 23:55:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ccec6900537c6c4243816dcd96b5fb9427f6bb5","subject":"Add Jammy refresh NS records","status":"MERGED","created":"2023-04-20 06:21:06.000000000","updated":"2023-04-25 23:57:30.000000000","submitted":"2023-04-25 23:55:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"880906-dns-jammy","meta_rev_id":"2646dbf14e29ead3ef17c097c1fe9129e111f7d6","_number":880906,"virtual_id_number":880906,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-25 23:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2023-04-25 23:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2023-04-20 20:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2023-04-25 23:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 06:23:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 20:41:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-04-20 20:43:00.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"b7192fd96b8eebc7a5afdf712e011a8b27c0b68e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-20 06:21:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6477c45d0082ccc87293465775a70ca8756608ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6121231aee204413a6189325e47ab739\n\n- validate-zone-db https://zuul.opendev.org/t/openstack/build/1ad717c07b814242ba6136cbc88699db : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ff603a366d1a5c9574da73923eaea1ed0e934824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-20 06:24:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"27e52817c61a492ae34798144103dcc2ef461305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:26:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be730fd0b9e241868e73f5278af62023\n\n- validate-zone-db https://zuul.opendev.org/t/openstack/build/c988396a7ddb41e592d3caa5c2febd8e : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8a6e53d0070d68c48f80f61d4cb0e326cb391153","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-20 20:41:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0f8c8ffe5b65bf2ce244af72dba8f8e61a73736","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2023-04-20 20:43:00.000000000","message":"Patch Set 2: Code-Review-1\n\nShould these servers be answering queries before we merge this?","accounts_in_message":[],"_revision_number":2},{"id":"63f2f0f5a17608dd891235a9e4c6a411e10663dd","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2023-04-25 23:38:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d7dbfc350ebe50007c43ac3295361bc50bc4a83f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:38:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9404c229ef9654823d0b5d7366eaa5c0c287da48","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2023-04-25 23:40:56.000000000","message":"Patch Set 2: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2924a6ff12da81c9e12d4443afd59a2babcf8b76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:41:09.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"0dfb96c0c7be46a41597af651f6e7312c4ffc33a","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2023-04-25 23:46:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6acd93ebb36631c8fabbcd7837a3e699c9666b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:49:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/011fc7a6ef78413b940c92ade3ab1866\n\n- validate-zone-db https://zuul.opendev.org/t/openstack/build/fa7a31c336cd48a09cdbc0fce0bc6c2f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c55b9d60235768b666c5d6c92d5e4b81f4fd1de8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:49:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5267e0e39729c2d35304f81a708d80813e696a8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:55:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dca1eee5f1448af90119fc1fd79cf6e\n\n- validate-zone-db https://zuul.opendev.org/t/openstack/build/aa122dca94fd453fb9b9bdd22fa2715e : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3eb9d43a006550bd397ff7a343df630bd1243bd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:55:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2646dbf14e29ead3ef17c097c1fe9129e111f7d6","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 23:57:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b555bbe51be4f5583d3b8a85c1af90f\n\n- infra-prod-service-nameserver https://zuul.opendev.org/t/openstack/build/906bc168abdc42ecb1e5a7fc43b344e7 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47b24489c25ae2f4d040291486fc1bd15cc42048","revisions":{"f4d87c2867406547c98ef8c36e2c3ebab7d08582":{"kind":"REWORK","_number":1,"created":"2023-04-20 06:21:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/06/880906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/zone-gating.dev","ref":"refs/changes/06/880906/1","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/1"}}},"commit":{"parents":[{"commit":"842ab4b335aad365290a111ac2769a7bfbc2b31f","subject":"Set default ttl to one hour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/zone-gating.dev/commit/842ab4b335aad365290a111ac2769a7bfbc2b31f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-20 05:58:00.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-20 06:21:01.000000000","tz":600},"subject":"Add Jammy refresh NS records","message":"Add Jammy refresh NS records\n\nAdd the Jammy refresh nameservers to the NS records.  This should be\ndone before updating the registry records.\n\nChange-Id: I1ccec6900537c6c4243816dcd96b5fb9427f6bb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/zone-gating.dev/commit/f4d87c2867406547c98ef8c36e2c3ebab7d08582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/zone-gating.dev/commit/f4d87c2867406547c98ef8c36e2c3ebab7d08582"}]},"branch":"refs/heads/master"},"47b24489c25ae2f4d040291486fc1bd15cc42048":{"kind":"REWORK","_number":2,"created":"2023-04-20 06:24:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/06/880906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/opendev/zone-gating.dev","ref":"refs/changes/06/880906/2","commands":{"Checkout":"git fetch https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/opendev/zone-gating.dev refs/changes/06/880906/2"}}},"commit":{"parents":[{"commit":"842ab4b335aad365290a111ac2769a7bfbc2b31f","subject":"Set default ttl to one hour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/zone-gating.dev/commit/842ab4b335aad365290a111ac2769a7bfbc2b31f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-20 05:58:00.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-20 06:22:49.000000000","tz":600},"subject":"Add Jammy refresh NS records","message":"Add Jammy refresh NS records\n\nAdd the Jammy refresh nameservers to the NS records.  This should be\ndone before updating the registry records.\n\nChange-Id: I1ccec6900537c6c4243816dcd96b5fb9427f6bb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/zone-gating.dev/commit/47b24489c25ae2f4d040291486fc1bd15cc42048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/opendev/zone-gating.dev/commit/47b24489c25ae2f4d040291486fc1bd15cc42048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
