)]}'
{"id":"openinfra%2Fopenstack-map~618111","triplet_id":"openinfra%2Fopenstack-map~master~I50470c07fd89cd2c09ca4febebdd30ebd17eb5d0","project":"openinfra/openstack-map","branch":"master","hashtags":[],"change_id":"I50470c07fd89cd2c09ca4febebdd30ebd17eb5d0","subject":"Add OCI","status":"ABANDONED","created":"2018-11-15 09:10:15.000000000","updated":"2020-05-04 11:30:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdecafaa7a191059dd8e9ce3a3ad41e5fab46992","_number":618111,"virtual_id_number":618111,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"date":"2019-03-17 21:44:36.000000000","_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":1,"date":"2018-11-15 09:10:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":-2,"date":"2019-11-27 15:13:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-15 09:10:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 15:13:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"85aab5fdb389bd2049cda9397a027a487d03beee","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2018-11-15 09:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f6c816170000d7a416e4b079f9c3550aea46e51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-15 09:10:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"3b0a1d6be0ed49f3fea62ef5cf90c06d144d8d35","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2018-11-29 16:37:00.000000000","message":"Patch Set 1:\n\nThanks for your proposed addition! We first need to decide how (if) we want to expose deployment tools that are not developed within the OpenStack community. Let\u0027s keep this review open while we figure this out.","accounts_in_message":[],"_revision_number":1},{"id":"544a1b4b28e88ab3942203ba2fd26a7103ec413a","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2018-12-06 15:32:21.000000000","message":"Patch Set 1:\n\nHi Thierry,\n\nIndeed, OCI isn\u0027t developed (yet) within the OpenStack community, however, it largely re-uses most of its components from it. All what it does is re-using puppet-openstack, which does the bulk of the work.\n\nHow can we move forward with this? Should we wait until the TC discusses it?","accounts_in_message":[],"_revision_number":1},{"id":"481ec84401ba49a8d3de9a6d5ce02f7a0db9ff78","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2018-12-07 13:16:06.000000000","message":"Patch Set 1:\n\nYes, ultimately that could be a TC decision, but the web team first needs to see what the design options are on the front-end side (separate tab, separate page etc). Please hold :)","accounts_in_message":[],"_revision_number":1},{"id":"83863e29fbd4ca264dcbd198c547304016a5ff68","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-02-15 15:53:23.000000000","message":"Patch Set 1: Workflow-1\n\nCurrent stance for openstack.org/software is to only list items officially produced by the OpenStack community. We plan to overhaul the deployers section (and the sdk section) though, in which case we might make room for 3rd-party tools. It\u0027s just not there yet.\n\nI propose that we keep this request open, to nudge us toward that direction. I\u0027ll set Workflow-1 to show that this is work in progress.","accounts_in_message":[],"_revision_number":1},{"id":"7c97b50f501499e1c239b211d8af721d5accbc71","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-03-17 21:44:36.000000000","message":"Patch Set 1:\n\nHi Thierry,\n\nIs there any progress here?\n\nIt would feel a little bit unfair if the software I wrote, using absolutely all of its bits from the OpenStack community (puppet-openstack, for example, to which I do contribute a fair amount of patches), had zero exposition in the openstack.org site.\n\nIf there is not the place to have such exposition, please let me know where. I would suggest just having another category if that\u0027s possible and desired. Maybe even 2:\n- free software, not maintain in OpenStack CI\n- proprietary software\n\nBy the way, I\u0027ve been pondering using the upstream CI for my deployment tool. This would probably mean a lot of involvement from my side though, and also maybe a lot of resources in the OpenStack CI, as it would need a lot of nodes to test the deployment. Maybe I\u0027ll open a thread about this in the general list.","accounts_in_message":[],"_revision_number":1},{"id":"135270071de25940bb1db4be3e1de3d2e43b3ed0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-03-18 10:17:43.000000000","message":"Patch Set 1:\n\nI agree that we could show open source OpenStack deployment tools that are not developed \"in\" OpenStack (under the TC) in a separate tab.\n\nIdeally we would first define a set of capabilities / tags that we would apply across the board, to help visitors understand what each solution actually allows to do. We have a session on that at the Denver forum:\n\nhttps://www.openstack.org/summit/denver-2019/summit-schedule/events/23682/deployment-tools-define-common-capabilities\n\nThat said, even if that session does not go anywhere, we are almost done with the backlog of improvements for the \"OpenStack components\" part of the software pages, so we should be able to work on improving the deployment tools soon. Adding a tab to allow displaying OCI and Kayobe is high on that list.","accounts_in_message":[],"_revision_number":1},{"id":"815ad29147add57f211123e880bef12dabd7124e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-27 15:13:44.000000000","message":"Patch Set 1: Code-Review-2 -Workflow\n\nAfter discussing it with the web team, I confirm that the current scope of the openstack.org/software section is only to list software directly produced by the OpenStack project, so we can\u0027t add third-party open source deployment solutions for now.\n\nThe place to list third-party solutions for deployment of OpenStack is the OpenStack marketplace (under the distributions and appliances section). If you\u0027re interested in getting OCI mentioned there, please contact danny@openstack.org who can help your project getting separately listed.\n\nTo facilitate finding those other deployment solutions, we will add a link to that section of the marketplace on the deployment tools page of the openstack project website.","accounts_in_message":[],"_revision_number":1},{"id":"bdecafaa7a191059dd8e9ce3a3ad41e5fab46992","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-04 11:30:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"554dc2fc0041503e0d341633ec43fca480e5ef76","revisions":{"554dc2fc0041503e0d341633ec43fca480e5ef76":{"kind":"REWORK","_number":1,"created":"2018-11-15 09:10:15.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/11/618111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/openstack-map","ref":"refs/changes/11/618111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/11/618111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/11/618111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/11/618111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/openstack-map refs/changes/11/618111/1"}}},"commit":{"parents":[{"commit":"243f76c38083393f5654c238bac0e7b4f6c26a21","subject":"Remove Dragonflow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/243f76c38083393f5654c238bac0e7b4f6c26a21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2018-11-15 09:09:34.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2018-11-15 09:10:10.000000000","tz":60},"subject":"Add OCI","message":"Add OCI\n\nChange-Id: I50470c07fd89cd2c09ca4febebdd30ebd17eb5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/554dc2fc0041503e0d341633ec43fca480e5ef76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/554dc2fc0041503e0d341633ec43fca480e5ef76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
