)]}'
{"id":"openinfra%2Fopenstack-map~940724","triplet_id":"openinfra%2Fopenstack-map~master~Ic9413061692d33aecf7d030641a527c6b53c420d","project":"openinfra/openstack-map","branch":"master","topic":"reintroduce-freezer","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2025-03-27 15:48:03.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}},"hashtags":[],"change_id":"Ic9413061692d33aecf7d030641a527c6b53c420d","subject":"Reintroduce Freezer","status":"MERGED","created":"2025-02-04 17:14:05.000000000","updated":"2025-03-27 15:50:29.000000000","submitted":"2025-03-27 15:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"940724-reintroduce-freezer","meta_rev_id":"3b9c6b1c77e0ad4d2ab33874b58f74ef1148deeb","_number":940724,"virtual_id_number":940724,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 15:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-03-27 15:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-03-27 14:04:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-03-27 15:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-04 17:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-07 14:14:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-03-27 14:04:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"09a8089e237de390564363b0e6b6bd5e829afc45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-02-04 17:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6955eef1479adccea897b669d716079781a2fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 17:16:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c36195cbf84931868181960ed08a6d\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b83cd640faa64483b99cb07d4fb05fed : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4b7a87d188274f9fb8612de539711274aeb85b32","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-07 14:14:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"451a31823cbee815c25efd9aa6ba54c3b068df63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-03-27 13:08:54.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f6e945e10cac4203c9d7fab64d531fec7271278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:46:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b55c40ac2eb54466b0094437bf471c10\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34855d1de383442ca7c31a719225a591 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e8557ce83abd29e9446fda4a924db658b6370821","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-27 14:04:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2f8748f8567fc228abe083aa61b2a4c8e77ad1a0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-03-27 15:48:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e8d22c81e13eaa30f2aee877ffee3a1e16a82543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:48:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57c395e9cf1e3b5c55ed65a880f99b59773f1f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:50:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8a33a83d68a40a19414b5e4c13afb25\n\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/624e560bb99646e7b0c653be5a2d2795 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3b9c6b1c77e0ad4d2ab33874b58f74ef1148deeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c43b687dc996d85e06cd7b13d00bad72aca3dc5","revisions":{"808bae9f47098572d4e16cf7241966f109d4991d":{"kind":"REWORK","_number":1,"created":"2025-02-04 17:14:05.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/24/940724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/openstack-map","ref":"refs/changes/24/940724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/1"}}},"commit":{"parents":[{"commit":"5593591e060f1cf6cc6ba7b6355a7acfb279ccc4","subject":"Add coordination tag to components","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/5593591e060f1cf6cc6ba7b6355a7acfb279ccc4"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2025-02-04 17:11:16.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2025-02-04 17:14:03.000000000","tz":60},"subject":"Reintroduce Freezer","message":"Reintroduce Freezer\n\nFreezer was removed in April 2024, using the following change:\nhttps://review.opendev.org/c/openinfra/openstack-map/+/914948\n\nThis patch reverts the Freezer-related part of this change, after\nFreezer was officially removed from the inactive projects list:\nhttps://review.opendev.org/c/openstack/governance/+/938938\n\nChange-Id: Ic9413061692d33aecf7d030641a527c6b53c420d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/808bae9f47098572d4e16cf7241966f109d4991d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/808bae9f47098572d4e16cf7241966f109d4991d"}]},"branch":"refs/heads/master"},"8c43b687dc996d85e06cd7b13d00bad72aca3dc5":{"kind":"REWORK","_number":2,"created":"2025-03-27 13:08:54.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/24/940724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/openstack-map","ref":"refs/changes/24/940724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/openstack-map refs/changes/24/940724/2"}}},"commit":{"parents":[{"commit":"5593591e060f1cf6cc6ba7b6355a7acfb279ccc4","subject":"Add coordination tag to components","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/5593591e060f1cf6cc6ba7b6355a7acfb279ccc4"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2025-02-04 17:11:16.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2025-03-27 13:08:54.000000000","tz":0},"subject":"Reintroduce Freezer","message":"Reintroduce Freezer\n\nFreezer was removed in April 2024, using the following change:\nhttps://review.opendev.org/c/openinfra/openstack-map/+/914948\n\nThis patch reverts the Freezer-related part of this change, after\nFreezer was officially removed from the inactive projects list:\nhttps://review.opendev.org/c/openstack/governance/+/938938\n\nChange-Id: Ic9413061692d33aecf7d030641a527c6b53c420d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/8c43b687dc996d85e06cd7b13d00bad72aca3dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstack-map/commit/8c43b687dc996d85e06cd7b13d00bad72aca3dc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
