)]}'
{"id":"openinfra%2Fopenstackid~709864","triplet_id":"openinfra%2Fopenstackid~master~Ic9dff4d7114b1866cc8156c79befabc1612157fa","project":"openinfra/openstackid","branch":"master","topic":"release-jobs","hashtags":[],"change_id":"Ic9dff4d7114b1866cc8156c79befabc1612157fa","subject":"Migrate openstackid-release-branch to Zuul v3","status":"MERGED","created":"2020-02-25 23:26:29.000000000","updated":"2020-02-25 23:52:22.000000000","submitted":"2020-02-25 23:48:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709864-1582674505386-b16e9fe7","meta_rev_id":"53d115d4a9a9ce42eb59d0409468321bcdc54e4d","_number":709864,"virtual_id_number":709864,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 23:48:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-02-25 23:33:26.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-02-25 23:39:35.000000000","_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-02-25 23:36:46.000000000","_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 23:33:26.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-25 23:36:46.000000000","updated_by":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"reviewer":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"state":"REVIEWER"},{"updated":"2020-02-25 23:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62d472b288a3d28a1381733de6b964bc046394e8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-25 23:26:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f00aa06f00f24cbac590bf1795a0179cb58cdb2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-25 23:33:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c736914ab53a245f24922bb14d6ba5f8cef8b472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:36:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstackid-unittests https://zuul.opendev.org/t/openstack/build/47a9bc824a454e63a399e78edd444f71 : SUCCESS in 8m 12s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1f15d7fc5d2f4d80aa4e156cd22ea6f9 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bd26449597aebff3652523f92d7c61df75ca6a7f","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-25 23:36:46.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0c387766f385f2c1f2baa433a59e18d3388e5cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:37:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17f920578071a569adf960788b061e080f35c165","author":{"_account_id":9139,"name":"sebastian marcet","email":"smarcet@gmail.com","username":"smarcet"},"date":"2020-02-25 23:39:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ede3e78287ab6a800474fc8f4d2557aa1624326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:48:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstackid-unittests https://zuul.opendev.org/t/openstack/build/9f5ebdc998fd4a2c819d968f97fa22e5 : SUCCESS in 7m 01s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e94226769aee4a23a6a97c0b0ae45a76 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d681834b5470ea6d0b5a511fa680bb999e85ce4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:48:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6e2c480f6d939372f489dd151e6b71cdcfbd679a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:52:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/e11d4a60b73043ce9b7d1606fb0564b5 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88c5d8511c38fdad4d3788a9abccb3981f8a987e","revisions":{"88c5d8511c38fdad4d3788a9abccb3981f8a987e":{"kind":"REWORK","_number":1,"created":"2020-02-25 23:26:29.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/64/709864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/openstackid","ref":"refs/changes/64/709864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/openstackid refs/changes/64/709864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/openstackid refs/changes/64/709864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/openstackid refs/changes/64/709864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/openstackid refs/changes/64/709864/1"}}},"commit":{"parents":[{"commit":"3d3a41b4946ddf46de94a5784c99bb09dc066696","subject":"Merge \"Also run openstackid-release-branch in release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstackid/commit/3d3a41b4946ddf46de94a5784c99bb09dc066696"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-02-25 23:25:20.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-02-25 23:25:20.000000000","tz":0},"subject":"Migrate openstackid-release-branch to Zuul v3","message":"Migrate openstackid-release-branch to Zuul v3\n\nGet rid of transitional legacy compat code in the playbooks for the\nopenstackid-release-branch job, and reparent it on the non-legacy\npublication job.\n\nChange-Id: Ic9dff4d7114b1866cc8156c79befabc1612157fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstackid/commit/88c5d8511c38fdad4d3788a9abccb3981f8a987e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/openstackid/commit/88c5d8511c38fdad4d3788a9abccb3981f8a987e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
