)]}'
{"id":"openinfra%2Fpython-tempestconf~581344","triplet_id":"openinfra%2Fpython-tempestconf~master~Idc03155b345d341903c2941f41ec4677a503b531","project":"openinfra/python-tempestconf","branch":"master","topic":"fix","hashtags":[],"change_id":"Idc03155b345d341903c2941f41ec4677a503b531","subject":"Fix http_image uploading to glance","status":"MERGED","created":"2018-07-10 12:07:30.000000000","updated":"2018-07-16 22:08:22.000000000","submitted":"2018-07-16 22:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"581344-1531778902529-a66f025f","meta_rev_id":"0753fdea04edb7638a4b2291d493429bf5913448","_number":581344,"virtual_id_number":581344,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"date":"2018-07-16 15:50:45.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2018-07-16 22:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-07-16 18:54:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2018-07-12 09:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2018-07-12 08:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"date":"2018-07-12 15:29:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2018-07-12 09:44:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-12 08:55:23.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2018-07-12 09:44:51.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2018-07-12 15:29:29.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2018-07-16 18:54:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-16 22:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"144a725439ee6f655adf4b72e935a06a4b963e56","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-10 12:07:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1539b6e01063c79ef3cbfa916a1306c3a5ef1341","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-10 12:07:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0ca40dbfe7edbee3c3d9028ef4afac12662666d0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-07-10 12:29:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"48eae3a1f82846a113240f08b5f81602e5c100a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-10 14:35:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/c9bec71/ : SUCCESS in 1h 53m 11s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/31b9c2a/ : SUCCESS in 1h 35m 34s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b1f7f2e/ : SUCCESS in 2h 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"9a3990ab818d6e058eb0539ccd49363846d97b28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 15:03:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/2/check/openstack-tox-pep8/7bc6398/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/2/check/openstack-tox-py27/b420f45/ : SUCCESS in 7m 42s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/2/check/openstack-tox-py35/f372fae/ : SUCCESS in 6m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/2/check/build-openstack-sphinx-docs/9242eef/html/ : SUCCESS in 5m 31s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/2/check/python-tempestconf-tox-cover/098e94d/ : SUCCESS in 5m 38s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/2/check/python-tempestconf-tempest-devstack-admin/f7785da/ : SUCCESS in 37m 48s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/2/check/python-tempestconf-tempest-devstack-demo/9429f4b/ : SUCCESS in 39m 04s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/2/check/python-tempestconf-tempest-packstack-admin/8a32bd8/ : SUCCESS in 1h 06m 06s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/2/check/python-tempestconf-tempest-packstack-demo/50d1ae6/ : SUCCESS in 52m 24s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/2/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/a94f33e/ : FAILURE in 2h 45m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c502be1fec48f93af18631512c0eafa3e3b05af2","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-10 15:03:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6fdada96fb6d2785afd7c0e00a10b6bf0a298ef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 16:41:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/730c961/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/0462e12/ : SUCCESS in 6m 14s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/70abd91/ : SUCCESS in 5m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/f1f9dd8/html/ : SUCCESS in 5m 14s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/dac501f/ : SUCCESS in 6m 27s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/cfc0647/ : SUCCESS in 41m 51s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/35f0845/ : SUCCESS in 37m 21s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/075d8de/ : SUCCESS in 1h 01m 41s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/c303fdd/ : SUCCESS in 59m 10s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/3219afd/ : RETRY_LIMIT in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"7943703678654560e299ae01a45e5d6c1d4d0ce6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-07-10 17:33:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e6792dff34da38fe1cb4421effdebf5dbfbdeac9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-10 17:54:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/dbf79ff/ : SUCCESS in 1h 54m 45s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/0fb99d5/ : SUCCESS in 1h 37m 18s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c3accb6/ : SUCCESS in 2h 10m 00s","accounts_in_message":[],"_revision_number":3},{"id":"634083b68b42ecb119a5a9f575ed489725f5edb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 20:59:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/1fab4b8/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/ed3b63f/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/bc42de0/ : SUCCESS in 5m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/fadd91b/html/ : SUCCESS in 5m 28s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/bae208b/ : SUCCESS in 5m 34s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/4b45658/ : SUCCESS in 37m 26s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/e8c24ab/ : SUCCESS in 33m 36s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/8c8e96e/ : SUCCESS in 1h 05m 23s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/79579bb/ : SUCCESS in 1h 07m 57s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/0e5d1e4/ : FAILURE in 3h 18m 01s","accounts_in_message":[],"_revision_number":3},{"id":"12082016d96323b7acb7d24574dfb509e100e4f1","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-10 21:08:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bbfef4ccf6bb54c2dfd89e1c3c68096a65d630a0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-10 23:18:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/32254f7/ : SUCCESS in 1h 49m 38s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/4c5cecf/ : SUCCESS in 1h 36m 01s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bdb8a5a/ : SUCCESS in 2h 07m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a48220237c2be92ceddc95f1ae771bdb6221595b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 23:57:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/54a6265/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/05699b4/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/3ac7ccf/ : SUCCESS in 5m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/9c43b33/html/ : SUCCESS in 5m 37s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/2a716a0/ : SUCCESS in 4m 52s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/96190b8/ : SUCCESS in 35m 24s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/80b9095/ : SUCCESS in 40m 49s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/36b5315/ : SUCCESS in 1h 38m 45s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/ce1ff2a/ : SUCCESS in 1h 03m 41s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/6a210d3/ : FAILURE in 2h 41m 42s","accounts_in_message":[],"_revision_number":3},{"id":"061992a1d91d2007d0b4d77861f8e1fb9681482c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-11 12:31:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"932acd40b110b6d04cce2ba8f71f199fd5eb272e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-11 14:38:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/02363fe/ : SUCCESS in 1h 47m 14s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/d933c6d/ : SUCCESS in 1h 33m 19s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/194cce4/ : SUCCESS in 2h 01m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d59e94e4a18c680f852edd2e0268079c3618fa14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 14:58:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/5d183f6/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/89ec6c7/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/904ff3f/ : SUCCESS in 7m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/c451350/html/ : SUCCESS in 4m 47s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/b35373c/ : SUCCESS in 4m 54s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/bbc44fd/ : SUCCESS in 37m 26s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/4a59621/ : SUCCESS in 36m 09s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/a064a1e/ : SUCCESS in 1h 04m 12s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/2feab79/ : SUCCESS in 1h 07m 59s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/4977d68/ : SUCCESS in 2h 25m 23s","accounts_in_message":[],"_revision_number":3},{"id":"a2a34be866139924efc5186304537eb78448a2bf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-07-11 15:22:27.000000000","message":"Patch Set 3:\n\nDo we need to set image.image_path in tempest.conf then? It\u0027s only used to upload the image to glance, and later tempest does not it anymore. Only image.http_image is relevant for tempest.","accounts_in_message":[],"_revision_number":3},{"id":"1a28ad8bf99361debdd5f054be3d0755f00d8a4e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-11 16:00:15.000000000","message":"Patch Set 3:\n\nimage.image_path is relevant only from python-tempestconf\u0027s view. I think, this solution is clearer than the other ones I have tried. For example I\u0027ve tried to set the image in set_image_preferences, but in that case, the image couldn\u0027t be used in set_default_tempest_options, because that method is called earlier (during the discovery, right after the Image class is instantiated)..\n\nAnd maybe it\u0027s better for debugging too, because a user can see in tempest.conf what image url/path was used to upload the image to glance.","accounts_in_message":[],"_revision_number":3},{"id":"48a7fe270461178c8b3b2072292d47ac7852c558","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-07-12 08:55:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"52d3d301b681684ce78c6f8f329287109b82fbd4","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-07-12 09:26:58.000000000","message":"Patch Set 3:\n\nI agree with the approach current solution. I\u0027m still not sure that this value should be stored at all in tempest.conf: it has no value in tempest.conf, and you can see the value used from the logs of tempestconf.\n\nShouldn\u0027t releasenotes/notes/allow-non-admin-user-upload-image-f2274cdec154a76b.yaml be changed?","accounts_in_message":[],"_revision_number":3},{"id":"ed398ffde35dc12cfb63e160195e38890d325714","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-07-12 09:44:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThat said, the configuration key can be removed from tempest.conf in another review.","accounts_in_message":[],"_revision_number":3},{"id":"a64a6bba35a178aa900a15993a617b492882e8a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 09:45:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e458c8df1b382d89dab2e5663508f7568744e04d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 12:08:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/gate/openstack-tox-pep8/59fa59e/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py27/54281ed/ : SUCCESS in 5m 31s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py35/9e7d0b1/ : SUCCESS in 7m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/gate/build-openstack-sphinx-docs/98b126d/html/ : SUCCESS in 4m 38s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-admin/17b185e/ : SUCCESS in 37m 43s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-demo/1e5c5a6/ : SUCCESS in 41m 07s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-admin/f9ce3ee/ : FAILURE in 1h 10m 29s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-demo/8c5f618/ : SUCCESS in 1h 04m 06s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/gate/tripleo-ci-centos-7-scenario002-multinode-oooq-container/9e89fec/ : FAILURE in 2h 19m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a0373caec18ff05b73d8816869e3727e16e3c759","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-07-12 12:10:48.000000000","message":"Patch Set 3:\n\nIs it safe to recheck, or is a more general failure of that job and it\u0027s better to wait?","accounts_in_message":[],"_revision_number":3},{"id":"968e05b51824c11f9874c38a139d498bf3a3223e","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-07-12 12:15:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a89b3ff457f34fe334e4487b9b29e45d76316482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 15:28:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/eb518d0/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/e2d02f0/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/c4ebc5f/ : SUCCESS in 4m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/1846c1d/html/ : SUCCESS in 3m 58s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/2253f50/ : SUCCESS in 4m 18s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/b5af668/ : SUCCESS in 36m 04s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/be7aa79/ : SUCCESS in 37m 23s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/529523b/ : SUCCESS in 1h 10m 03s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/b79d42a/ : SUCCESS in 1h 12m 11s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/b2c853c/ : POST_FAILURE in 2h 52m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8da860e8dd89bf7419abf7cd3cbb215a51ceaabd","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-07-12 15:28:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"54e9c60c0537f1d4e5a198220cd8f8d90ba46d64","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2018-07-12 15:29:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4f6d1682696ce0c448af2e8b02488c16e681fbfc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-12 16:22:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/318a899/ : SUCCESS in 1h 44m 09s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/d56bc35/ : FAILURE in 16m 56s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7eeaafa/ : FAILURE in 1h 12m 08s","accounts_in_message":[],"_revision_number":3},{"id":"7ac40d20ba9279438931daedb1af2831d31a70b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 18:52:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/7c11648/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/82d7183/ : SUCCESS in 6m 30s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/f3f1a0b/ : SUCCESS in 7m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/d541f73/html/ : SUCCESS in 4m 53s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/be40aac/ : SUCCESS in 5m 30s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/549fefe/ : SUCCESS in 36m 40s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/8e9c392/ : SUCCESS in 37m 47s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/6e87a06/ : SUCCESS in 1h 02m 14s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/3634022/ : SUCCESS in 1h 00m 00s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/35d4183/ : SUCCESS in 2h 15m 55s","accounts_in_message":[],"_revision_number":3},{"id":"817288c6ef4f81601d44c8874535618c5716f903","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 18:52:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"becf1f5e937f82d9c163846ea4dcd2097209c905","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 21:51:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/gate/openstack-tox-pep8/04e5b68/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py27/4d15521/ : SUCCESS in 7m 26s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py35/0c084d9/ : SUCCESS in 5m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/gate/build-openstack-sphinx-docs/b22fd5d/html/ : SUCCESS in 5m 28s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-admin/7af41a9/ : SUCCESS in 37m 42s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-demo/5231267/ : SUCCESS in 36m 55s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-admin/83ca5c5/ : SUCCESS in 1h 00m 22s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-demo/e2c591e/ : SUCCESS in 1h 00m 04s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/gate/tripleo-ci-centos-7-scenario002-multinode-oooq-container/06b88eb/ : FAILURE in 2h 53m 36s","accounts_in_message":[],"_revision_number":3},{"id":"d69f346df91dd73d768b2f6c8672cb5ea42e25dc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-13 08:18:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c5b53962491ffdce54d6495f805c8cb78f4318a0","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-07-13 08:30:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"57aabc061ff79d63edd34f293f9def161924e0a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-13 11:03:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/e08bc99/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/878114c/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/a06b83b/ : SUCCESS in 4m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/4568210/html/ : SUCCESS in 4m 04s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/5d52572/ : SUCCESS in 5m 10s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/052f1f7/ : SUCCESS in 35m 13s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/e0e396e/ : SUCCESS in 29m 27s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/1309e36/ : SUCCESS in 1h 06m 32s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/93e91f7/ : SUCCESS in 59m 38s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/9fd76cb/ : FAILURE in 2h 44m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1519779750a9bfb57cfc3267fedd478865728773","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-13 11:16:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/2f5da09/ : SUCCESS in 1h 46m 01s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/0d076b7/ : FAILURE in 17m 51s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1f876cc/ : SUCCESS in 2h 00m 17s","accounts_in_message":[],"_revision_number":3},{"id":"963a878d3645a2b0995fc22155b80ce06aa43667","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-07-16 09:40:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cde1411d7e4821658b2922cb696e70b84f7e58a0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-16 11:44:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/efdb665/ : SUCCESS in 1h 59m 57s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/805fc24/ : SUCCESS in 1h 31m 48s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1573ffb/ : FAILURE in 2h 00m 31s","accounts_in_message":[],"_revision_number":3},{"id":"01893b5e22447ef340a1e8c211e38e0d5b55526c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 12:33:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/d975069/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/08f2080/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/83bc3b2/ : SUCCESS in 5m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/f9a9146/html/ : SUCCESS in 6m 37s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/51325a8/ : SUCCESS in 5m 26s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/3624144/ : SUCCESS in 35m 04s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/a46d060/ : SUCCESS in 31m 58s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/1fb939d/ : SUCCESS in 1h 06m 24s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/f981a79/ : SUCCESS in 58m 01s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/715618e/ : SUCCESS in 2h 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"737058f0af1ece1434c1deb64134f592891c8405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 12:34:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5824ca07d6e12b76f32ce6afe98a23675a5edabe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 15:40:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/gate/openstack-tox-pep8/c22809c/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py27/9d7ff8c/ : SUCCESS in 7m 24s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py35/98b955b/ : SUCCESS in 5m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/gate/build-openstack-sphinx-docs/2fa11fa/html/ : SUCCESS in 4m 37s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-admin/7fe770b/ : SUCCESS in 34m 38s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-demo/c111e2b/ : SUCCESS in 35m 08s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-admin/840f7db/ : SUCCESS in 1h 04m 58s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-demo/4dc2b3d/ : SUCCESS in 57m 29s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/gate/tripleo-ci-centos-7-scenario002-multinode-oooq-container/5561094/ : TIMED_OUT in 3h 00m 59s","accounts_in_message":[],"_revision_number":3},{"id":"93d30ac0e9027552e345c72962c2f45b56491048","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2018-07-16 15:50:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0fe7110fb48d05f4cedfa1b0584151dd389a859d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-16 18:54:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/cdfeb26/ : SUCCESS in 1h 47m 02s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/51dd8d2/ : SUCCESS in 1h 27m 08s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/44/581344/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e8a3dbf/ : SUCCESS in 1h 53m 04s","accounts_in_message":[],"_revision_number":3},{"id":"e1d5d0e3f8f97d3be21dce713893f35e2a145be5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 19:41:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/check/openstack-tox-pep8/f6c9b34/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/check/openstack-tox-py27/7b89843/ : SUCCESS in 6m 15s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/check/openstack-tox-py35/56bf056/ : SUCCESS in 6m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/check/build-openstack-sphinx-docs/f335629/html/ : SUCCESS in 5m 39s\n- python-tempestconf-tox-cover http://logs.openstack.org/44/581344/3/check/python-tempestconf-tox-cover/e0eaaa9/ : SUCCESS in 5m 35s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-admin/b98f4a0/ : SUCCESS in 35m 02s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-devstack-demo/5c666f0/ : SUCCESS in 32m 46s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-admin/28346ce/ : SUCCESS in 1h 08m 40s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/check/python-tempestconf-tempest-packstack-demo/920bbce/ : SUCCESS in 1h 02m 56s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/48121a7/ : SUCCESS in 2h 20m 37s","accounts_in_message":[],"_revision_number":3},{"id":"21ac01450226ab7f3eb0f2e26a8f688cd0bacb79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 19:41:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea425230fc01cc7fa3f72431dd9f6cb978d0f53a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 22:08:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/581344/3/gate/openstack-tox-pep8/65a443b/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py27/875b58b/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/44/581344/3/gate/openstack-tox-py35/c1bd95b/ : SUCCESS in 5m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/44/581344/3/gate/build-openstack-sphinx-docs/c158f4a/html/ : SUCCESS in 5m 20s\n- python-tempestconf-tempest-devstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-admin/48cfdfa/ : SUCCESS in 34m 17s\n- python-tempestconf-tempest-devstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-devstack-demo/1604521/ : SUCCESS in 35m 06s\n- python-tempestconf-tempest-packstack-admin http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-admin/fba7a23/ : SUCCESS in 1h 03m 35s\n- python-tempestconf-tempest-packstack-demo http://logs.openstack.org/44/581344/3/gate/python-tempestconf-tempest-packstack-demo/b91c5d3/ : SUCCESS in 1h 03m 08s\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/44/581344/3/gate/tripleo-ci-centos-7-scenario002-multinode-oooq-container/c4ae6c5/ : SUCCESS in 2h 20m 49s","accounts_in_message":[],"_revision_number":3},{"id":"53f387c52b0bf7a80965137311db29b26244e890","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-16 22:08:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dcc4e9a5ddec55b4c79b69df2d3f2ed7a0e4d985","revisions":{"26ed983af5412d2248f235db9c74f6c29d8f822d":{"kind":"REWORK","_number":1,"created":"2018-07-10 12:07:30.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/44/581344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/44/581344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/1"}}},"commit":{"parents":[{"commit":"188c145281451df5cd693abf6e777278b25ce319","subject":"Document installation process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/188c145281451df5cd693abf6e777278b25ce319"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2018-07-10 11:52:40.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2018-07-10 12:03:10.000000000","tz":0},"subject":"Fix http_image uploading to glance","message":"Fix http_image uploading to glance\n\nimage.http_image was during the refactoring mistaken for an image\nwhich is uploaded to glance and that caused the issues described\nin the linked BZ.\nimage.http_image is a tempest option which defines http accessible\nimage. It can be also in a compressed format, so it can\u0027t be\nuploaded to glance before it\u0027s unpacked first.\nThe patch introduces a new python-tempestconf option in tempest.conf\ndefining an image which will be uploaded to glance.\n\nChange-Id: Idc03155b345d341903c2941f41ec4677a503b531\nBZ:1599372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/26ed983af5412d2248f235db9c74f6c29d8f822d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/26ed983af5412d2248f235db9c74f6c29d8f822d"}]},"branch":"refs/heads/master"},"47d55bd42e348f7b806d8959580dfa3d8fc62ca8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-07-10 12:07:56.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/44/581344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/44/581344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/2"}}},"commit":{"parents":[{"commit":"3a40d5fe982b0b8ef49f3b16bfbc278a10a93ef0","subject":"Replace tripleo job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/3a40d5fe982b0b8ef49f3b16bfbc278a10a93ef0"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2018-07-10 11:52:40.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2018-07-10 12:04:22.000000000","tz":0},"subject":"Fix http_image uploading to glance","message":"Fix http_image uploading to glance\n\nimage.http_image was during the refactoring mistaken for an image\nwhich is uploaded to glance and that caused the issues described\nin the linked BZ.\nimage.http_image is a tempest option which defines http accessible\nimage. It can be also in a compressed format, so it can\u0027t be\nuploaded to glance before it\u0027s unpacked first.\nThe patch introduces a new python-tempestconf option in tempest.conf\ndefining an image which will be uploaded to glance.\n\nChange-Id: Idc03155b345d341903c2941f41ec4677a503b531\nBZ:1599372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/47d55bd42e348f7b806d8959580dfa3d8fc62ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/47d55bd42e348f7b806d8959580dfa3d8fc62ca8"}]},"branch":"refs/heads/master"},"dcc4e9a5ddec55b4c79b69df2d3f2ed7a0e4d985":{"kind":"REWORK","_number":3,"created":"2018-07-10 15:03:40.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/44/581344/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/44/581344/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/44/581344/3"}}},"commit":{"parents":[{"commit":"3a40d5fe982b0b8ef49f3b16bfbc278a10a93ef0","subject":"Replace tripleo job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/3a40d5fe982b0b8ef49f3b16bfbc278a10a93ef0"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2018-07-10 11:52:40.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2018-07-10 14:59:37.000000000","tz":0},"subject":"Fix http_image uploading to glance","message":"Fix http_image uploading to glance\n\nimage.http_image was during the refactoring mistaken for an image\nwhich is uploaded to glance and that caused the issues described\nin the linked story.\nimage.http_image is a tempest option which defines http accessible\nimage. It can be also in a compressed format, so it can\u0027t be\nuploaded to glance before it\u0027s unpacked first.\nThe patch introduces a new python-tempestconf option in tempest.conf\ndefining an image which will be uploaded to glance.\n\nStory: 2002931\nTask: 22918\nChange-Id: Idc03155b345d341903c2941f41ec4677a503b531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/dcc4e9a5ddec55b4c79b69df2d3f2ed7a0e4d985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/dcc4e9a5ddec55b4c79b69df2d3f2ed7a0e4d985"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
