)]}'
{"id":"openinfra%2Fpython-tempestconf~675105","triplet_id":"openinfra%2Fpython-tempestconf~master~I891b1721c625c61880f289e74ad09d54df9c6f7a","project":"openinfra/python-tempestconf","branch":"master","topic":"os_client_config_update","hashtags":[],"change_id":"I891b1721c625c61880f289e74ad09d54df9c6f7a","subject":"Add Queens, Rocky and Stein job","status":"MERGED","created":"2019-08-07 12:25:02.000000000","updated":"2019-08-09 11:52:18.000000000","submitted":"2019-08-09 11:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675105-1565351538308-fcaf60fb","meta_rev_id":"28c75df4c7dd91e0f80c218152155c172b693d56","_number":675105,"virtual_id_number":675105,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-08-09 11:52:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-08 16:46:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2019-08-08 22:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-09 10:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 10:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 16:46:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-08 22:07:33.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-08-09 10:10:27.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-08-09 11:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11c1bd008d1b5610f65403e2aaf65ad1caeede3b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 12:25:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ac27a0c64732320cd99ad6b8fd7ae0ed2c4e1e2","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 12:48:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2399bcd90d658e66895642423bcd001438f1b037","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:46:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/05/675105/2/check/openstack-tox-cover/695791c/cover/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/2/check/openstack-tox-pep8/1f14d53/ : SUCCESS in 3m 35s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/2/check/openstack-tox-py27/af3be29/ : SUCCESS in 4m 43s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/2/check/openstack-tox-py35/4b3e17b/ : SUCCESS in 4m 50s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/2/check/openstack-tox-py36/539c44f/ : SUCCESS in 5m 01s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/2/check/openstack-tox-py37/4c1c94d/ : SUCCESS in 3m 27s\n- openstack-tox-docs https://logs.opendev.org/05/675105/2/check/openstack-tox-docs/f3f078a/html/ : SUCCESS in 4m 08s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin/5a17837/ : SUCCESS in 47m 55s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin-rocky/ec1d010/ : SUCCESS in 54m 35s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin-queens/6247e81/ : SUCCESS in 49m 33s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin-py3/eb68af5/ : SUCCESS in 49m 56s\n- python-tempestconf-tempest-devstack-admin-py3-rocky https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin-py3-rocky/09b4ab5/ : FAILURE in 57m 05s\n- python-tempestconf-tempest-devstack-admin-py3-queens https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin-py3-queens/09c299d/ : FAILURE in 38m 36s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-demo/2df8b9f/ : SUCCESS in 47m 17s\n- python-tempestconf-tempest-devstack-demo-rocky https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-demo-rocky/a1c82e3/ : SUCCESS in 1h 05m 48s\n- python-tempestconf-tempest-devstack-demo-queens https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-demo-queens/e684ea5/ : SUCCESS in 41m 36s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-demo-py3/4c850ab/ : SUCCESS in 51m 10s\n- python-tempestconf-tempest-devstack-demo-py3-rocky https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-demo-py3-rocky/12953c0/ : FAILURE in 34m 37s\n- python-tempestconf-tempest-devstack-demo-py3-queens https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-demo-py3-queens/0a203b9/ : FAILURE in 36m 33s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-packstack-admin/ee69482/ : SUCCESS in 1h 18m 29s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-packstack-demo/b414834/ : SUCCESS in 1h 09m 38s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/2/check/python-tempestconf-tempest-devstack-admin-plugins/2d97b9b/ : SUCCESS in 50m 01s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/2/check/tripleo-ci-centos-7-scenario002-standalone/ebe0f14/ : SUCCESS in 1h 54m 50s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/2/check/tripleo-ci-centos-7-standalone-os-tempest/dabd110/ : SUCCESS in 1h 16m 41s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/2/check/refstack-client-devstack-tempestconf/51756f0/ : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2cfc8e61531aabb57fc1489a17372ab36168680c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 15:34:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/675105/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e6b69fc/ : SUCCESS in 2h 44m 34s","accounts_in_message":[],"_revision_number":2},{"id":"64c1464368258588f61e86cbe5aac5c8560dd7a0","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 16:58:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ed8b5f171cecbaa9b63b58c67433a59f6273de8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 18:32:06.000000000","message":"Patch Set 3:\n\nhmm:\n\n2019-08-07 17:33:42.537292 | LOOP [generate-tempestconf-file : Create python-tempestconf venv with latest pip, setuptools and pbr]\n2019-08-07 17:33:43.571077 | controller | ERROR: Item: [\u0027pip\u0027, \u0027setuptools\u0027, \u0027pbr\u0027]\n2019-08-07 17:33:43.571848 | controller | {\n2019-08-07 17:33:43.571969 | controller |   \"item\": [\n2019-08-07 17:33:43.572078 | controller |     \"pip\",\n2019-08-07 17:33:43.572184 | controller |     \"setuptools\",\n2019-08-07 17:33:43.572315 | controller |     \"pbr\"\n2019-08-07 17:33:43.572461 | controller |   ],\n2019-08-07 17:33:43.572690 | controller |   \"msg\": \"parameters are mutually exclusive: name, requirements\"\n2019-08-07 17:33:43.572856 | controller | }\n2019-08-07 17:33:43.595273 | \n\nAnd pip ansible module doens\u0027t support constraints parameter yet:\nhttps://github.com/ansible/ansible/issues/47654","accounts_in_message":[],"_revision_number":3},{"id":"f85cfe8b2707ec56f704a6973d83b681fbafe23d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 18:41:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fc2c6a60b2b5033f6f5fda81abfef69484413f68","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-07 19:31:54.000000000","message":"Patch Set 4:\n\nNot experimental?","accounts_in_message":[],"_revision_number":4},{"id":"73db021e81c73b95974f819370be3490d644e294","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 20:34:20.000000000","message":"Patch Set 4:\n\n\u003e Not experimental?\n\nI don\u0027t know, per the discussion in the dependent review, the jobs should be permanent or experimental .. speaking about that, what do you actually mean by experimental?","accounts_in_message":[],"_revision_number":4},{"id":"1f180929a95a643195691f4125fc67f8d208892d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-07 20:34:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a823ddd20164cc8e6445dd87e5eebbd1c7ec01e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 22:15:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/05/675105/5/check/openstack-tox-cover/5474227/cover/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/5/check/openstack-tox-pep8/d0259a9/ : SUCCESS in 4m 09s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/5/check/openstack-tox-py27/a4416a3/ : SUCCESS in 4m 15s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/5/check/openstack-tox-py35/6b74050/ : SUCCESS in 3m 56s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/5/check/openstack-tox-py36/29a3b5a/ : SUCCESS in 3m 53s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/5/check/openstack-tox-py37/005fe94/ : SUCCESS in 3m 40s\n- openstack-tox-docs https://logs.opendev.org/05/675105/5/check/openstack-tox-docs/54a9913/html/ : SUCCESS in 3m 33s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-admin/dc0e479/ : SUCCESS in 1h 00m 20s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-admin-rocky/b649774/ : FAILURE in 41m 57s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-admin-queens/e487581/ : FAILURE in 32m 23s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-admin-py3/b7579a8/ : SUCCESS in 45m 04s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-demo/2a61cd1/ : SUCCESS in 44m 54s\n- python-tempestconf-tempest-devstack-demo-rocky https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-demo-rocky/af6c2d0/ : FAILURE in 34m 49s\n- python-tempestconf-tempest-devstack-demo-queens https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-demo-queens/1e67130/ : FAILURE in 38m 37s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-demo-py3/1dfa74b/ : SUCCESS in 44m 10s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-packstack-admin/125009e/ : SUCCESS in 1h 24m 45s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-packstack-demo/3765b4e/ : SUCCESS in 1h 12m 14s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/5/check/python-tempestconf-tempest-devstack-admin-plugins/2c64f04/ : SUCCESS in 52m 24s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/5/check/tripleo-ci-centos-7-scenario002-standalone/7c4406f/ : SUCCESS in 1h 39m 53s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/5/check/tripleo-ci-centos-7-standalone-os-tempest/cda7571/ : SUCCESS in 1h 21m 00s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/5/check/refstack-client-devstack-tempestconf/416e0f8/ : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":5},{"id":"a01fa1a6b17470ff3b0ec9208ac13e4f65366c26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 23:34:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/675105/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3c6565a/ : SUCCESS in 2h 59m 22s","accounts_in_message":[],"_revision_number":5},{"id":"369a46ce22e46c2e26078f0cdade649c13a3d7b0","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-08 05:09:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"75ce8f40923f670869cc91a0a9353f7c8edb8a78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 06:51:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/05/675105/6/check/openstack-tox-cover/e316aa8/cover/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/6/check/openstack-tox-pep8/92220e3/ : SUCCESS in 3m 05s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/6/check/openstack-tox-py27/acb92c1/ : SUCCESS in 3m 26s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/6/check/openstack-tox-py35/bf23f40/ : SUCCESS in 3m 19s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/6/check/openstack-tox-py36/99fd864/ : SUCCESS in 3m 19s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/6/check/openstack-tox-py37/03ff1c4/ : SUCCESS in 3m 27s\n- openstack-tox-docs https://logs.opendev.org/05/675105/6/check/openstack-tox-docs/8372e15/html/ : SUCCESS in 2m 45s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-admin/509e398/ : SUCCESS in 58m 05s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-admin-rocky/d4a30aa/ : FAILURE in 34m 42s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-admin-queens/61b954d/ : FAILURE in 33m 01s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-admin-py3/63a647e/ : SUCCESS in 47m 13s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-demo/927b8dc/ : SUCCESS in 51m 41s\n- python-tempestconf-tempest-devstack-demo-rocky https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-demo-rocky/19becc0/ : FAILURE in 33m 22s\n- python-tempestconf-tempest-devstack-demo-queens https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-demo-queens/97e441c/ : FAILURE in 32m 26s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-demo-py3/6edfd43/ : SUCCESS in 39m 04s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-packstack-admin/fe7cddd/ : FAILURE in 6m 59s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-packstack-demo/355d791/ : FAILURE in 4m 13s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/6/check/python-tempestconf-tempest-devstack-admin-plugins/9c07151/ : SUCCESS in 52m 25s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/6/check/tripleo-ci-centos-7-scenario002-standalone/b695d79/ : SUCCESS in 1h 38m 52s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/6/check/tripleo-ci-centos-7-standalone-os-tempest/4f18e75/ : FAILURE in 45m 08s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/6/check/refstack-client-devstack-tempestconf/a647418/ : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":6},{"id":"6b0c6ce0e5f7a15e3143fb2648c74590f43d9563","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 07:47:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/675105/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8754db3/ : SUCCESS in 2h 37m 33s","accounts_in_message":[],"_revision_number":6},{"id":"0413cd1a1ec23daf096ec309c3fa44817769edc2","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-08 08:50:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8e73d5f9d467465dc23efb71414e601adac41c8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:32:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/05/675105/7/check/openstack-tox-cover/34b5c28/cover/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/7/check/openstack-tox-pep8/300cd63/ : SUCCESS in 3m 50s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/7/check/openstack-tox-py27/33ebbe3/ : SUCCESS in 4m 20s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/7/check/openstack-tox-py35/2473dec/ : SUCCESS in 5m 59s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/7/check/openstack-tox-py36/dafbda0/ : SUCCESS in 3m 31s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/7/check/openstack-tox-py37/3468db4/ : SUCCESS in 4m 11s\n- openstack-tox-docs https://logs.opendev.org/05/675105/7/check/openstack-tox-docs/3daefb2/html/ : SUCCESS in 4m 05s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-admin/eac416d/ : SUCCESS in 46m 31s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-admin-rocky/f6ac048/ : SUCCESS in 50m 02s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-admin-queens/f459cc5/ : SUCCESS in 55m 16s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-admin-py3/cdda53b/ : SUCCESS in 50m 17s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-demo/bb76474/ : SUCCESS in 41m 35s\n- python-tempestconf-tempest-devstack-demo-rocky https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-demo-rocky/5fae539/ : SUCCESS in 43m 01s\n- python-tempestconf-tempest-devstack-demo-queens https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-demo-queens/5e7a6ed/ : SUCCESS in 42m 51s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-demo-py3/43ac283/ : SUCCESS in 38m 41s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-packstack-admin/41227f4/ : SUCCESS in 1h 15m 03s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-packstack-demo/aea42e4/ : SUCCESS in 1h 13m 03s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/7/check/python-tempestconf-tempest-devstack-admin-plugins/9e16989/ : SUCCESS in 52m 03s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/7/check/tripleo-ci-centos-7-scenario002-standalone/618207f/ : SUCCESS in 1h 33m 46s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/7/check/tripleo-ci-centos-7-standalone-os-tempest/96cfd99/ : SUCCESS in 1h 15m 14s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/7/check/refstack-client-devstack-tempestconf/cb74f21/ : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":7},{"id":"d85a636db9c411a2ce53231c2c0de445b36af7fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 11:59:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/675105/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f001965/ : SUCCESS in 3h 09m 01s","accounts_in_message":[],"_revision_number":7},{"id":"77f4ad55deb81016adfe072adbb17ab60dc7e8ba","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-08 13:10:15.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"53a6360e2069f0da17fcc13a85935a7bbec16c92","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-08 13:33:31.000000000","message":"Patch Set 7: Code-Review-1\n\nWe need also the Stein jobs.\n\nBut then we would be adding 6 new jobs at once.\n\nMaybe we can add just the -admin jobs for the 3 old branches, and add the -demo jobs to the experimental queue. That should allow us to catch issues and not kill the queue.","accounts_in_message":[],"_revision_number":7},{"id":"21d2b9e65872200e71a3f0b6e3b05a7e717aa46f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-08 13:50:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"15e6f41759856d031adab4be6a74f9f462053ff2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-08 15:00:08.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"e901e2b9b067e64d96dffd97d12775b2413d0792","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:55:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- python-tempestconf-tempest-devstack-demo-stein https://logs.opendev.org/05/675105/8/experimental/python-tempestconf-tempest-devstack-demo-stein/1568bcd/ : SUCCESS in 48m 27s\n- python-tempestconf-tempest-devstack-demo-rocky https://logs.opendev.org/05/675105/8/experimental/python-tempestconf-tempest-devstack-demo-rocky/1ee17bc/ : SUCCESS in 43m 50s\n- python-tempestconf-tempest-devstack-demo-queens https://logs.opendev.org/05/675105/8/experimental/python-tempestconf-tempest-devstack-demo-queens/989a45f/ : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":8},{"id":"0af79b747f7bd1f1da487a6461f89af6920ca2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:55:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/05/675105/8/check/openstack-tox-cover/bd13913/cover/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/8/check/openstack-tox-pep8/51fef49/ : SUCCESS in 3m 29s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/8/check/openstack-tox-py27/9cac428/ : SUCCESS in 3m 37s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/8/check/openstack-tox-py35/70b4bc8/ : SUCCESS in 6m 51s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/8/check/openstack-tox-py36/2f3ca8b/ : SUCCESS in 4m 42s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/8/check/openstack-tox-py37/cb10db3/ : SUCCESS in 4m 47s\n- openstack-tox-docs https://logs.opendev.org/05/675105/8/check/openstack-tox-docs/74d603b/html/ : SUCCESS in 3m 41s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin/a9d4a52/ : SUCCESS in 49m 59s\n- python-tempestconf-tempest-devstack-admin-stein https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-stein/017d838/ : SUCCESS in 57m 44s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-rocky/bf19383/ : SUCCESS in 1h 03m 02s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-queens/978b281/ : SUCCESS in 1h 00m 05s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-py3/39ba86e/ : SUCCESS in 53m 45s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-demo/b94c357/ : SUCCESS in 48m 41s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-demo-py3/dcc7014/ : SUCCESS in 48m 33s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-packstack-admin/bfc1350/ : TIMED_OUT in 2h 01m 37s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-packstack-demo/3751ada/ : SUCCESS in 1h 01m 36s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-plugins/d5eb7d4/ : SUCCESS in 1h 01m 43s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/8/check/tripleo-ci-centos-7-scenario002-standalone/36ad351/ : SUCCESS in 1h 58m 31s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/8/check/tripleo-ci-centos-7-standalone-os-tempest/070a829/ : SUCCESS in 1h 29m 28s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/8/check/refstack-client-devstack-tempestconf/5a2aa7b/ : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":8},{"id":"ca8bc4a5cd49a0bb336c55f6ef5ff2846ddd5fde","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-08 16:26:19.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e3fba299106ec4015926347ef519fd8c9632a149","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 16:46:40.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/675105/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/88383c2/ : FAILURE in 2h 54m 29s","accounts_in_message":[],"_revision_number":8},{"id":"2f9dfc96b700a89d88ec29c6a2ee509c837866fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:29:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/05/675105/8/check/openstack-tox-cover/4049415/cover/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/8/check/openstack-tox-pep8/d9d2767/ : SUCCESS in 3m 48s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/8/check/openstack-tox-py27/6909625/ : SUCCESS in 4m 38s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/8/check/openstack-tox-py35/c1a1caf/ : SUCCESS in 3m 47s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/8/check/openstack-tox-py36/f65df61/ : SUCCESS in 4m 34s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/8/check/openstack-tox-py37/1a8dc53/ : SUCCESS in 4m 47s\n- openstack-tox-docs https://logs.opendev.org/05/675105/8/check/openstack-tox-docs/da755ac/html/ : SUCCESS in 2m 52s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin/16e81c6/ : SUCCESS in 56m 59s\n- python-tempestconf-tempest-devstack-admin-stein https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-stein/c400800/ : SUCCESS in 1h 03m 09s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-rocky/5c490b0/ : SUCCESS in 56m 34s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-queens/a07785f/ : SUCCESS in 50m 36s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-py3/ff7cec4/ : SUCCESS in 49m 26s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-demo/e622411/ : SUCCESS in 49m 49s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-demo-py3/d94d43f/ : SUCCESS in 42m 49s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-packstack-admin/266b521/ : SUCCESS in 1h 34m 17s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-packstack-demo/5713234/ : SUCCESS in 1h 15m 26s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/8/check/python-tempestconf-tempest-devstack-admin-plugins/6e45bce/ : SUCCESS in 50m 14s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/8/check/tripleo-ci-centos-7-scenario002-standalone/5245e70/ : SUCCESS in 1h 26m 37s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/8/check/tripleo-ci-centos-7-standalone-os-tempest/9ff5f3f/ : SUCCESS in 1h 07m 11s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/8/check/refstack-client-devstack-tempestconf/2306987/ : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":8},{"id":"cf26f03243263aa05a03c046b009f696cdf2e1ca","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-08 22:07:33.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"103f8ca65d6bd61044c750b659c8b9a03514cfb3","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-09 10:10:17.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3b8343bb6d43d2755fcab590c412d4ca7aa136b2","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-09 10:10:27.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"fbdabbe6483e539064c0c96b6ab2ea4a9bdf3c2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 10:10:41.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3dda46fc3cdece8a54892387b2d999902ec69bdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 11:52:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"511a275342c5af4804e19f4c964ba3b60838ffdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 11:52:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/05/675105/8/gate/openstack-tox-pep8/1b2cc3a/ : SUCCESS in 3m 56s\n- openstack-tox-py27 https://logs.opendev.org/05/675105/8/gate/openstack-tox-py27/9100394/ : SUCCESS in 3m 40s\n- openstack-tox-py35 https://logs.opendev.org/05/675105/8/gate/openstack-tox-py35/77af415/ : SUCCESS in 3m 30s\n- openstack-tox-py36 https://logs.opendev.org/05/675105/8/gate/openstack-tox-py36/f147317/ : SUCCESS in 3m 54s\n- openstack-tox-py37 https://logs.opendev.org/05/675105/8/gate/openstack-tox-py37/ba383a5/ : SUCCESS in 3m 00s\n- openstack-tox-docs https://logs.opendev.org/05/675105/8/gate/openstack-tox-docs/edf370d/html/ : SUCCESS in 2m 26s\n- python-tempestconf-tempest-devstack-admin https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-admin/dc6350e/ : SUCCESS in 51m 49s\n- python-tempestconf-tempest-devstack-admin-stein https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-admin-stein/d634b51/ : SUCCESS in 49m 06s\n- python-tempestconf-tempest-devstack-admin-rocky https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-admin-rocky/8e853f7/ : SUCCESS in 52m 47s\n- python-tempestconf-tempest-devstack-admin-queens https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-admin-queens/e8fd2ab/ : SUCCESS in 52m 48s\n- python-tempestconf-tempest-devstack-admin-py3 https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-admin-py3/f553763/ : SUCCESS in 51m 18s\n- python-tempestconf-tempest-devstack-demo https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-demo/1cc545e/ : SUCCESS in 44m 52s\n- python-tempestconf-tempest-devstack-demo-py3 https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-demo-py3/abb6e21/ : SUCCESS in 43m 22s\n- python-tempestconf-tempest-packstack-admin https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-packstack-admin/711433d/ : SUCCESS in 1h 11m 35s\n- python-tempestconf-tempest-packstack-demo https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-packstack-demo/0cecc15/ : SUCCESS in 1h 13m 26s\n- python-tempestconf-tempest-devstack-admin-plugins https://logs.opendev.org/05/675105/8/gate/python-tempestconf-tempest-devstack-admin-plugins/72abef8/ : SUCCESS in 1h 01m 47s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/05/675105/8/gate/tripleo-ci-centos-7-scenario002-standalone/008c3ca/ : SUCCESS in 1h 32m 58s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/05/675105/8/gate/tripleo-ci-centos-7-standalone-os-tempest/bd2137c/ : SUCCESS in 1h 17m 47s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/05/675105/8/gate/refstack-client-devstack-tempestconf/a6f882d/ : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d4b7e15123cc50faae64f4110051f9d2a6c902bb","revisions":{"9906289b0b52a8bc09176c5dd7c2b4f3c4c9a9cd":{"kind":"REWORK","_number":1,"created":"2019-08-07 12:25:02.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/1"}}},"commit":{"parents":[{"commit":"e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c","subject":"Replace os-client-config module by openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"subject":"DNM testing queens and rocky stable branches","message":"DNM testing queens and rocky stable branches\n\nTesting changes of the dependent review against queens and rocky\nstable branches of devstack as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/9906289b0b52a8bc09176c5dd7c2b4f3c4c9a9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/9906289b0b52a8bc09176c5dd7c2b4f3c4c9a9cd"}]},"branch":"refs/heads/master"},"2f480db16fc8d414d3d2c8dd3229427ef2763565":{"kind":"REWORK","_number":2,"created":"2019-08-07 12:48:42.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/2"}}},"commit":{"parents":[{"commit":"e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c","subject":"Replace os-client-config module by openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:48:06.000000000","tz":0},"subject":"DNM testing queens and rocky stable branches","message":"DNM testing queens and rocky stable branches\n\nTesting changes of the dependent review against queens and rocky\nstable branches of devstack as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/2f480db16fc8d414d3d2c8dd3229427ef2763565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/2f480db16fc8d414d3d2c8dd3229427ef2763565"}]},"branch":"refs/heads/master"},"eefe93c61eceb9ae04d1f1c9ccef481eb5f62b11":{"kind":"REWORK","_number":3,"created":"2019-08-07 16:58:13.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/3"}}},"commit":{"parents":[{"commit":"e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c","subject":"Replace os-client-config module by openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 16:55:51.000000000","tz":0},"subject":"Add Queens and Rocky job","message":"Add Queens and Rocky job\n\nLet\u0027s test python-tempestconf patches with older releases as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/eefe93c61eceb9ae04d1f1c9ccef481eb5f62b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/eefe93c61eceb9ae04d1f1c9ccef481eb5f62b11"}]},"branch":"refs/heads/master"},"55c13f3db9781a1b7eca6fdbdea6281b1a307a2f":{"kind":"REWORK","_number":4,"created":"2019-08-07 18:41:00.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/4"}}},"commit":{"parents":[{"commit":"e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c","subject":"Replace os-client-config module by openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 18:39:58.000000000","tz":0},"subject":"Add Queens and Rocky job","message":"Add Queens and Rocky job\n\nLet\u0027s test python-tempestconf patches with older releases as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/55c13f3db9781a1b7eca6fdbdea6281b1a307a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/55c13f3db9781a1b7eca6fdbdea6281b1a307a2f"}]},"branch":"refs/heads/master"},"5134d6588c237a21206f67bd9b7710892def30f0":{"kind":"REWORK","_number":5,"created":"2019-08-07 20:34:32.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/5"}}},"commit":{"parents":[{"commit":"e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c","subject":"Replace os-client-config module by openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 20:31:50.000000000","tz":0},"subject":"Add Queens and Rocky job","message":"Add Queens and Rocky job\n\nLet\u0027s test python-tempestconf patches with older releases as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/5134d6588c237a21206f67bd9b7710892def30f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/5134d6588c237a21206f67bd9b7710892def30f0"}]},"branch":"refs/heads/master"},"b94471e94865b4865a161314e3468562e890790a":{"kind":"REWORK","_number":6,"created":"2019-08-08 05:09:13.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/6"}}},"commit":{"parents":[{"commit":"e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c","subject":"Replace os-client-config module by openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e1c61e9ec5750f32a26d99fd6e74e36f5881dc6c"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-08 05:08:40.000000000","tz":0},"subject":"Add Queens and Rocky job","message":"Add Queens and Rocky job\n\nLet\u0027s test python-tempestconf patches with older releases as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b94471e94865b4865a161314e3468562e890790a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b94471e94865b4865a161314e3468562e890790a"}]},"branch":"refs/heads/master"},"0c765be7e6c53efd2aed3f14048eb65715d21403":{"kind":"REWORK","_number":7,"created":"2019-08-08 08:50:04.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/7"}}},"commit":{"parents":[{"commit":"01f2ab1ea471e42588a595e7cc7336a58bafa1d3","subject":"Don\u0027t fail when heat-plugin is missing a value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/01f2ab1ea471e42588a595e7cc7336a58bafa1d3"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-08 08:49:30.000000000","tz":0},"subject":"Add Queens and Rocky job","message":"Add Queens and Rocky job\n\nLet\u0027s test python-tempestconf patches with older releases as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/0c765be7e6c53efd2aed3f14048eb65715d21403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/0c765be7e6c53efd2aed3f14048eb65715d21403"}]},"branch":"refs/heads/master"},"d4b7e15123cc50faae64f4110051f9d2a6c902bb":{"kind":"REWORK","_number":8,"created":"2019-08-08 13:50:48.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/675105/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/05/675105/8","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/05/675105/8"}}},"commit":{"parents":[{"commit":"01f2ab1ea471e42588a595e7cc7336a58bafa1d3","subject":"Don\u0027t fail when heat-plugin is missing a value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/01f2ab1ea471e42588a595e7cc7336a58bafa1d3"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-07 12:23:10.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-08 13:49:23.000000000","tz":0},"subject":"Add Queens, Rocky and Stein job","message":"Add Queens, Rocky and Stein job\n\nLet\u0027s test python-tempestconf patches with older releases as well.\n\nChange-Id: I891b1721c625c61880f289e74ad09d54df9c6f7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/d4b7e15123cc50faae64f4110051f9d2a6c902bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/d4b7e15123cc50faae64f4110051f9d2a6c902bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
