)]}'
{"id":"openinfra%2Fpython-tempestconf~713165","triplet_id":"openinfra%2Fpython-tempestconf~master~Ic35661c2795c3f12605696d4f29119a43608bf9b","project":"openinfra/python-tempestconf","branch":"master","topic":"centos-8-jobs","hashtags":[],"change_id":"Ic35661c2795c3f12605696d4f29119a43608bf9b","subject":"Switch to c8 based tripleo standalone jobs","status":"MERGED","created":"2020-03-16 06:12:09.000000000","updated":"2020-03-16 11:06:48.000000000","submitted":"2020-03-16 11:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713165-1584356582977-3ef7b696","meta_rev_id":"3c9c5825096c40e7929815cf82338829897189eb","_number":713165,"virtual_id_number":713165,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-16 11:03:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-16 08:51:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-03-16 09:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-16 09:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"date":"2020-03-16 06:14:36.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-03-16 06:14:42.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"date":"2020-03-16 06:14:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-03-16 09:10:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-16 06:14:36.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-03-16 06:14:42.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-03-16 06:14:52.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-03-16 08:51:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-16 09:06:48.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-03-16 09:10:31.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-03-16 11:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5c5eea0208c9df2bb04a9c3292045ddbb92fc4f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-03-16 06:12:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47fef7cb3cb984193e266404b72e02cd9bccc9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 08:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c765c2055ccf43c481d82c7c90c86152 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b771370d738c40d1b306f7d77380b115 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2bbc032c2b44775928bd3cd6be4ee49 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5fc250fdab84232b23d4c6c2b9e4c01 : SUCCESS in 6m 00s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/ae37ed23e58343238d8c90dbb990fdeb : SUCCESS in 49m 57s\n- python-tempestconf-tempest-devstack-admin-train https://zuul.opendev.org/t/openstack/build/b123e03f81a9465cb8277e84232cfef7 : SUCCESS in 1h 00m 21s\n- python-tempestconf-tempest-devstack-admin-stein https://zuul.opendev.org/t/openstack/build/8912965106264d5587bdacfa185bb699 : SUCCESS in 59m 46s\n- python-tempestconf-tempest-devstack-admin-rocky https://zuul.opendev.org/t/openstack/build/60d96076860e4b968d622303dde33922 : SUCCESS in 49m 40s\n- python-tempestconf-tempest-devstack-admin-queens https://zuul.opendev.org/t/openstack/build/45993c68d11a4d478ff57dc858677d2f : SUCCESS in 56m 56s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/2dcca086da944c6a92bfcb126ddfc729 : SUCCESS in 46m 51s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/e6ea8bb8af094e5da7f9a58fbb8c252c : SUCCESS in 53m 59s\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/6ede17dec8854441ba23037ba299f320 : SUCCESS in 56m 18s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/6d6f2d587fc448d18bc3215c7e2dd64f : SUCCESS in 1h 08m 14s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b17af9495ce347e4a95e4b54829e5e0a : SUCCESS in 2h 04m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/237a3058d6a94725b7c49fa5fa92a084 : SUCCESS in 1h 42m 59s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/d6d38b2344c44ae19cc10be3d311cbc4 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"dc6c2e292f9de6eb0d9381fa9bb7f87d258c33f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-16 08:51:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f6ea23c9c02b4620bae06a07a47cb311 : FAILURE in 2h 38m 04s","accounts_in_message":[],"_revision_number":1},{"id":"750528e65e4bafcb5d8007cc3e6a852382900968","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-03-16 09:06:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81042177e23262a4a8abcaec1f1babed7b7438ce","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-16 09:10:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0ed8dc9a7075ed55e3c394bc1e5b291d2e8de9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 09:10:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed8f1168d28c02bfd2d2f41ca93d8b72b8c402a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 11:03:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e3d50d22750c4b8b65a7cafcd973a20acd329b8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 11:03:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91540f55d841422da8be42f77a7f16a8 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a3850e63de047cc885d9b482aa3cebc : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1852594e8ced4962992464ccc6511ea0 : SUCCESS in 4m 04s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/db3e9705d6af46fb8de953d788615422 : SUCCESS in 45m 29s\n- python-tempestconf-tempest-devstack-admin-train https://zuul.opendev.org/t/openstack/build/8af726c42eac4f77a6a7c06362dcf652 : SUCCESS in 50m 22s\n- python-tempestconf-tempest-devstack-admin-stein https://zuul.opendev.org/t/openstack/build/0ba552d144324ef8825af137b5f9a98d : SUCCESS in 45m 31s\n- python-tempestconf-tempest-devstack-admin-rocky https://zuul.opendev.org/t/openstack/build/cb670a73d7234d5f8008116bd6a75365 : SUCCESS in 52m 20s\n- python-tempestconf-tempest-devstack-admin-queens https://zuul.opendev.org/t/openstack/build/021b7ba0bd8a4dcea232dca09bb0b9cc : SUCCESS in 44m 47s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/8628367bb25d492f9abcb2e1eb1bb9ff : SUCCESS in 46m 57s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/b34aa53888b44ab5accd4b6d083cd6ec : SUCCESS in 55m 57s\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/aef3aa094e4b43e7a6f7bccf28978020 : SUCCESS in 51m 38s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/53964ff4bb8f467b9a0d582a7ebbd9d0 : SUCCESS in 54m 56s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9ab29003383f44ffb2177fd65fa6d32d : SUCCESS in 1h 45m 36s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cc8c9063a4664632b87658a46d763cda : SUCCESS in 1h 31m 17s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/20877558af514d09b8e1670206a0ff95 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":1},{"id":"66f289061f6503d8cfa3754c593f53a00f01e200","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 11:06:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c70fcf3064074587b0e9259e0ed12a2f : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a267acfbece8be73567140143d67457fe58d2e7c","revisions":{"a267acfbece8be73567140143d67457fe58d2e7c":{"kind":"REWORK","_number":1,"created":"2020-03-16 06:12:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/65/713165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/65/713165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/65/713165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/65/713165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/65/713165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/65/713165/1"}}},"commit":{"parents":[{"commit":"c6cf6795143be999ec7813fa48612b736b1cb7f7","subject":"Added Mistral service type workflowv2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/c6cf6795143be999ec7813fa48612b736b1cb7f7"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-16 06:10:23.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-16 06:11:52.000000000","tz":330},"subject":"Switch to c8 based tripleo standalone jobs","message":"Switch to c8 based tripleo standalone jobs\n\nChange-Id: Ic35661c2795c3f12605696d4f29119a43608bf9b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/a267acfbece8be73567140143d67457fe58d2e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/a267acfbece8be73567140143d67457fe58d2e7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
