)]}'
{"id":"openinfra%2Fpython-tempestconf~730635","triplet_id":"openinfra%2Fpython-tempestconf~master~I54cdde0511e1579056fb08f01af0d323af20357f","project":"openinfra/python-tempestconf","branch":"master","topic":"dnm","hashtags":[],"change_id":"I54cdde0511e1579056fb08f01af0d323af20357f","subject":"DNM","status":"ABANDONED","created":"2020-05-25 15:11:42.000000000","updated":"2020-05-26 11:40:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c6e826923f9b211a6d30e0216b7899528cc84f6","_number":730635,"virtual_id_number":730635,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-25 17:17:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-25 16:39:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 16:39:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-25 17:17:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4972618a352995f1597b696fb4f75d24813a27b","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-05-25 15:11:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88e063a0d5735a1ee98f1ddcd87595db16849a5b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-25 16:39:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9dc722bc2c184398ab6fa80aac69f7e7 : POST_FAILURE in 1h 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c6443514970492f594f1973c11575230b8dc41a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 17:17:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51148c8033084f48a47596d11789b644 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/513fdf9f8b7247479645c9d93741127c : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b780b326ae15492fa1f7a429c328dccc : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8eb6de7a9a446fbb693cb8b5343caac : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783b140be8054db8bfb2cd18920f01ed : SUCCESS in 4m 03s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/85e3b34b2e8f433c8042ed4542adef89 : SUCCESS in 43m 41s\n- python-tempestconf-tempest-devstack-admin-train https://zuul.opendev.org/t/openstack/build/6509b9dc3cd449d092c8ee46231d0ada : SUCCESS in 47m 44s\n- python-tempestconf-tempest-devstack-admin-stein https://zuul.opendev.org/t/openstack/build/e63b131cc04b4eaa8ac7cba68beb0629 : SUCCESS in 48m 51s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/fd856a5fdb2849e094b7c1b7a6c604ed : SUCCESS in 45m 12s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/35578d64f5c44bfc90ebe55893af3805 : FAILURE in 29m 03s\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/b841a050999640d3ba37688395dbb7ba : FAILURE in 31m 54s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/ab77ba27ff8647f881e330b0ecad17d7 : SUCCESS in 52m 12s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/280dc2edc55944a3868d1d1a63c69530 : SUCCESS in 2h 04m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/efd5e60e291b4048a0c5f5f03d365d49 : SUCCESS in 1h 59m 18s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/164a54312abb4046b0e03536d6cedfa4 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0442041c03c022c6d03a3cb207bed3e859858069","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-05-26 11:40:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e2e180ec49117b09972aa83278e2eb086b1ddce","revisions":{"9e2e180ec49117b09972aa83278e2eb086b1ddce":{"kind":"REWORK","_number":1,"created":"2020-05-25 15:11:42.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/35/730635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/35/730635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/35/730635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/35/730635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/35/730635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/35/730635/1"}}},"commit":{"parents":[{"commit":"e5f2ba7c7d08d369f831409c3e07357d28d5feb9","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e5f2ba7c7d08d369f831409c3e07357d28d5feb9"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-05-25 15:11:12.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-05-25 15:11:12.000000000","tz":0},"subject":"DNM","message":"DNM\n\nChange-Id: I54cdde0511e1579056fb08f01af0d323af20357f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/9e2e180ec49117b09972aa83278e2eb086b1ddce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/9e2e180ec49117b09972aa83278e2eb086b1ddce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
