)]}'
{"id":"openinfra%2Fpython-tempestconf~870968","triplet_id":"openinfra%2Fpython-tempestconf~master~Ie8d0cd289302b789d48c925c1ab0b9940a6e09ac","project":"openinfra/python-tempestconf","branch":"master","topic":"fix/tox-docs","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-02-13 08:37:04.000000000","reason":"\u003cGERRIT_ACCOUNT_30674\u003e replied on the change","reason_account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}}},"hashtags":[],"change_id":"Ie8d0cd289302b789d48c925c1ab0b9940a6e09ac","subject":"Remove skipped test from designate-tempest-plugin","status":"MERGED","created":"2023-01-18 15:49:09.000000000","updated":"2023-02-13 11:53:11.000000000","submitted":"2023-02-13 11:51:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870968-fix/tox-docs","meta_rev_id":"fbb940eb49b8cba87d363c5bf901275c6cd42311","_number":870968,"virtual_id_number":870968,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 11:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2023-02-09 13:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-02-09 13:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 19:01:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-09 13:24:17.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"e7f2eb0aaa7ee46b5e8d1593da71d61ac72a36f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-18 15:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d97f798ff8f303910b2d5cff9aa17ef60a28a702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-18 15:58:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fb10c004f413094af274f8338362103eb5e75415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:01:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5e2028f66e74e108b043352a58d4d5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/091ece410dd4434a945a112426e5250d : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff616cc88bad4eb48442e54e5d5a72d7 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7503b2a52b4c41f8bdf5439286024b2a : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f890ef4ef3924636ac277f2ca2ade46b : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/750a9ba8f1db4e7c8a6e8a9860bd61f1 : SUCCESS in 2m 56s (non-voting)\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/811a39f4c65640e5a8bb933c02a2566a : SUCCESS in 1h 03m 38s\n- python-tempestconf-tempest-devstack-admin-victoria https://zuul.opendev.org/t/openstack/build/c813ad8ec8954dcf8528e5c221a81101 : FAILURE in 55m 46s\n- python-tempestconf-tempest-devstack-admin-wallaby https://zuul.opendev.org/t/openstack/build/1f2134a6bdf440e6b43e1c1e45fba77b : FAILURE in 45m 21s\n- python-tempestconf-tempest-devstack-admin-xena https://zuul.opendev.org/t/openstack/build/24643464cb3843ae80f124a8e29fa828 : SUCCESS in 53m 47s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/ce4ad2e611664be4ac846a544ec66eb4 : FAILURE in 49m 12s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/a8ae5ace66ab4ce6ae482feeb96f320f : SUCCESS in 55m 13s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/e8f5854ac1444d9fb52ad19f1a9849c9 : SUCCESS in 1h 17m 08s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/7c3f98138f93409f95cc568714d16a28 : SUCCESS in 47m 20s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/97ce1acdb1ae4418bfc368d4c4111fa9 : SUCCESS in 42m 49s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/a79ce2af90b440b59b6679820337a81c : SUCCESS in 1h 23m 53s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1fa409cfc40a45a4a0b49057c1871e0c : SUCCESS in 2h 56m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a99ddb486ae14afda37a2476aa6e0f5a : SUCCESS in 1h 30m 03s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/67b83f421b9c499aa6c5badfc666ef03 : FAILURE in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1f3f85a5da74b8c9033e36d08d855d39dee5d8d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-08 12:45:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c2e7908405906351ec44b07d1203618ac0228b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:37:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd6bfb27993c4575a9c26a7e8040076c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/021e7e549d6846d9bc0e136a9158226a : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea8a5764edf4fa8971d02c314605900 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e0696eeeb0b4e10ba83184d1a130a8f : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d601e9b548034b2294fe94c39b7a0d78 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b935f86f7534a0da92bbd793a0257d4 : SUCCESS in 3m 00s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/4cd4bcfe408c47d68934cfc567a3f218 : SUCCESS in 57m 21s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/d5f14d29e9eb46558d5b9234c9ae4a8c : SUCCESS in 49m 35s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/cd7f761483254f3280fbb5f7478346b4 : FAILURE in 1h 11m 53s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/b4f33cc2f553476ca1885a5d4def6ce9 : SUCCESS in 50m 39s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/972a71fb91a5460986a78575a6128e71 : SUCCESS in 1h 33m 39s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/6d7bdf4316cd4335b94beb80e0162627 : SUCCESS in 40m 45s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/03edf469d41c464e9f4bda1ca7b401b5 : SUCCESS in 1h 25m 13s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/2c4a30ea21a24f0c98bf4f91933fd421 : SUCCESS in 1h 18m 50s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ab9699998cbb478e89cf67e06ee57b1c : SUCCESS in 2h 43m 24s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1767069bafa54942b3f44c43591f24b3 : FAILURE in 1h 21m 42s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d921600eaa5b4e669829214ccfbc6f31 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"eac60e96803c1dce9fb81a263506757fb2a09bbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-09 11:06:36.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c6425ba278ec2d35c4675ea616b311956b6fdbcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 12:58:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69e419e0d3c74dc6a27ec3c8b84c3b15\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c10808fb37534cbab93506ce0a0c0ba6 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897b4d73f348431ea5257d177d6d31a7 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0046215f9ec49cba51aa0ec972896ae : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be1aed8b79154ba497eb3dc606337efa : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a19e32c0fbff4a0f835205abfcdb925f : SUCCESS in 2m 49s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/663776942eed4b85b6b289bf25b100a4 : SUCCESS in 57m 15s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/989060ff16214081ba0895365eb15a83 : SUCCESS in 48m 15s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/a2e6a629c9534abfabc26b0f1f15b8fd : FAILURE in 45m 14s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/95bee268741d4ceb91abdbe4a7937fbc : SUCCESS in 52m 02s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/4d65a4d3ecfc4b97bb9dce405da1a388 : SUCCESS in 1h 26m 26s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/e6deaacfe907441f8caceeaf67052177 : SUCCESS in 1h 07m 03s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/80f2330ad3d04ad2a09809d8d3098141 : SUCCESS in 36m 05s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/d72981b8e9b54016b7b71bf53cc8b185 : SUCCESS in 1h 26m 17s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1a9f2e90737941b99be77e65f52e4f76 : SUCCESS in 1h 12m 12s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/8e7d830016e947009f4cb88b8e006d36 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"3a7d89a86ca7db93f4dbe916a6c56b61c6c2c327","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-09 13:24:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ee5cf7db07fe9308b8f8442249db029e7d981fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 13:25:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aff39b9ee04a51f5b44afc018fbb718b131300bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:02:59.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/862818 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"75ddd58192035da3f9697f562924830399e64a6f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-10 13:06:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a7b201f41e965f4930726827b197e55dae34aee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 15:19:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2092e0b9b6b74fceab5895d23dd39cd1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee8474a2cc5645f1aa13cee1de8ce5e1 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3abc8e487be9431e8962b96d9d3dc9d0 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ecb2e7e0b3d4a8d8a58754c43058f0d : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/263ca7880f6c46d2958a8166122f9b0b : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c793664e6374a6b9f792aba664acaf3 : SUCCESS in 3m 17s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/37e4576d6afd4bbd913ae81e02ef534c : SUCCESS in 1h 00m 09s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/1c60e35fd1844b1a95238a3f1f287b20 : SUCCESS in 1h 11m 19s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/8ba00baa0b314dfa9d9758f2acb2306a : FAILURE in 49m 11s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/06c6e4f0ce5a472aad4aae172e6d4b30 : SUCCESS in 54m 35s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/09cfc749eedf46a8a58880395976a30a : SUCCESS in 1h 07m 05s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/69979f26a2f74edc83ba11230aa76c39 : SUCCESS in 46m 35s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/8e04a6b776064a5aa36bf55955109d80 : SUCCESS in 30m 50s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/7d73ffbcd0154588b3730db7362fa0cc : TIMED_OUT in 2h 03m 44s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0228b954e7c54246b607d0c2e956a701 : RETRY_LIMIT in 13s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/08cb626a85954cdc9996c2c19f5df6e6 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a430b2673cc4be03320af5f3430a0d15b17f93ab","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-10 15:27:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1227bbda8ec093e40f82d09614a76812b4a6a663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 17:58:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8788fed9c694666b6d0facb22221735\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20be36db75af4dc086cd0618e7608798 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f3b7d3db21e4c08b88e03e38465405c : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc5ed583e4ad4b86ab07bdc2a0692ef2 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c90c2102d7b4842813f0d52c8ed1e7d : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9a2327f004a47a68ca73d204fbe97bb : SUCCESS in 4m 51s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/e592ed62db174ed295770d2e4a9b2e85 : SUCCESS in 49m 10s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/9308912b9dd24974934acd24d012d72b : SUCCESS in 45m 14s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/2df66fece472433ca6d6c616df737959 : FAILURE in 42m 11s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/89a7cd35468b49ddb5e28d108dce9edc : SUCCESS in 49m 48s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/d453a0de25d64e9fb83f1ac5c4a4e2e5 : SUCCESS in 1h 17m 01s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/daf9ab8abd824534878657a96954147f : SUCCESS in 27m 09s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/e7310ffcd73f429bb2f7d7b4f39a7296 : SUCCESS in 32m 48s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/f35fdc4cc0ec458aac6d6219fdffc1dc : SUCCESS in 1h 14m 37s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b4722f9a072d496aa582310f4f462d5c : SUCCESS in 1h 04m 58s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3814f6b1b7cc40679a300d05d680f068 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"7e9a75acd01cdc87b12321193901051d79d9dd6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:04:00.000000000","message":"Patch Set 4: Verified-2\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":4},{"id":"af1bffd74156875b91f86ece78913b3e64a5a155","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-10 18:07:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fecbcd7c16c4c73c51435803463bf8e5b3351504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:07:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a12cb48163074380a9fce619acac0558\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04fc87924c904adeb345da2d3415d120 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2984f062cf54a3686934452c182d45e : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48314fa8acb94f2d91053d0687174528 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e4b642be360477899aa45fea66cb0e1 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b441ff7a877c4bc8be9b2a213b01f964 : SUCCESS in 3m 01s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/2e8d5d8ac9ba4458bc063bcd67f33442 : SUCCESS in 58m 10s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/60128bf8bb254f488ecfe1284d7d7f56 : SUCCESS in 58m 04s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/651292d8bfda40d2910c36e6c7b76ecf : FAILURE in 49m 24s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/cd68520afc144ec39b4728c8da2b58f8 : SUCCESS in 52m 11s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/b1817e8432be4c089deb7ec5035f522a : SUCCESS in 1h 25m 36s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/ef1a098f81f3464d8c6a8d4926e8e447 : SUCCESS in 36m 55s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/7ef27ef135774803a4f1b2e0d5ff6d63 : SUCCESS in 42m 55s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/a3ce6981063647b0a031c4bf307570fe : SUCCESS in 1h 23m 44s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3e4f43c599c4452fbe434a0216ae27cb : SUCCESS in 1h 47m 49s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4ee92fb574eb42eea86225ee953318aa : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"40b28956b262193550b51d0301018ee28ee1d61b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:07:40.000000000","message":"Patch Set 4: Verified-2\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":4},{"id":"70a9a36de1f691c6c4e201fe9491a5907eec4bc9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-11 08:38:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c966bf789e2e99d730405e62a4a4c1eaeb344de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-11 08:39:08.000000000","message":"Patch Set 4: Verified-1\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":4},{"id":"daa89a2b23b49f2c0af08344a192b998827508d2","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-13 08:37:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a27ae4b981dbf1ea2bbc7a4dbfbd64ac83b4e345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:15:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef496642a5f46df804b41f855a951f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c064bc87b1aa41c480230f26d1aa9df3 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe722840d7c94b45b11735afbe6b2442 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9199b6e9ba14550824ab60dcdfedd1f : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3828bacf3aa4df48d89e59bb188a104 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4046302aa1a745fd9c1f49e46c040ed9 : SUCCESS in 3m 29s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/3e5b10dd9d084fb2abd54860783b76e3 : SUCCESS in 50m 21s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/1922a6ed7452460cb00b228a40a22e12 : SUCCESS in 1h 12m 26s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/376b04bd8d1b47f4bb0c178e5b7cd039 : FAILURE in 41m 10s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/15389c7f6e7e45fcbbd21cca1b355a9a : SUCCESS in 33m 42s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/8dcd5e4aa55c4e768577e39b2fe90205 : SUCCESS in 1h 02m 42s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/2bc39e2e030444c584cd487e1f725dfe : SUCCESS in 43m 57s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/3488066b384041b08b118d438a57a4e7 : SUCCESS in 29m 01s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/d86f1e3d5701442f98fd2b24f5dc91f7 : SUCCESS in 1h 29m 45s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/31f7209b276943508d1e7674c7809170 : SUCCESS in 50m 54s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e779b8f2bda94321b3ee9ff5f214b32c : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"6280f1c67c453e11ede45764084873861157a629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:17:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c4aaf61139515cfc5a8a2b71ce2b93c118d26101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:51:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d964ea551e044946ab44ff0c80d31f50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e8669d9cc54a22bad40b4143c64590 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68801b3ff1354adfafc8d6ca1ffae25a : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41d87387476645fa8b113f47976a4ccb : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a63c44f00f247cc83ef9903ec8f9437 : SUCCESS in 2m 59s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/0f00ee75accf460ebb2cfe98d19d433e : SUCCESS in 35m 46s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/d6ab214ced8148dab915e8127c1c5bbd : SUCCESS in 40m 17s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/44d7ddf716424a348cedcc245d340efa : SUCCESS in 48m 44s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/8088e7c5b4014545b109fdcb39816aa7 : SUCCESS in 1h 06m 04s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/52e0d11340634a369de1288dd2247cac : SUCCESS in 1h 05m 57s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2a58e46094f94a4a92cc630e823ca035 : SUCCESS in 1h 24m 12s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0fea0630eb8849999c58743233b8a684 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"fe4ffbadd966859a3bc2533157eff0a987a97afc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:51:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fbb940eb49b8cba87d363c5bf901275c6cd42311","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:53:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bf93944f065461ab3c2ab8dd7611334\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/e1bb839782844a7db267c06e818c8a8a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5bc9166bcb459254141292e09389b3ee1becf1c8","revisions":{"d266789d2d5efcd9beca8e372df4b65e4eff4ffc":{"kind":"REWORK","_number":1,"created":"2023-01-18 15:49:09.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/68/870968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/68/870968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/1"}}},"commit":{"parents":[{"commit":"6488be3e8294ddceea0028b208bdcda791dc0a67","subject":"Change class variables to instance variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/6488be3e8294ddceea0028b208bdcda791dc0a67"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-18 15:44:41.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-18 15:49:07.000000000","tz":60},"subject":"Remove skipped test from designate-tempest-plugin","message":"Remove skipped test from designate-tempest-plugin\n\nWe can remove test_manually_trigger_update_secondary_zone_negative\ntest from exclude_regex as bug 968717 was fixed by removing the\ntest from the plugin [1].\n\n[1] https://bugs.launchpad.net/designate/+bug/1968717\n\nChange-Id: Ie8d0cd289302b789d48c925c1ab0b9940a6e09ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/d266789d2d5efcd9beca8e372df4b65e4eff4ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/d266789d2d5efcd9beca8e372df4b65e4eff4ffc"}]},"branch":"refs/heads/master"},"fb71022b3c085473a395585829fa87b54d3c8c9c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-18 15:58:54.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/68/870968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/68/870968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/2"}}},"commit":{"parents":[{"commit":"72f983bf53b93a8831acd96661b050ebcd741784","subject":"Return designate tests to admin-plugins job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/72f983bf53b93a8831acd96661b050ebcd741784"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-18 15:44:41.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-18 15:58:35.000000000","tz":60},"subject":"Remove skipped test from designate-tempest-plugin","message":"Remove skipped test from designate-tempest-plugin\n\nWe can remove test_manually_trigger_update_secondary_zone_negative\ntest from exclude_regex as bug 968717 was fixed by removing the\ntest from the plugin [1].\n\n[1] https://bugs.launchpad.net/designate/+bug/1968717\n\nChange-Id: Ie8d0cd289302b789d48c925c1ab0b9940a6e09ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/fb71022b3c085473a395585829fa87b54d3c8c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/fb71022b3c085473a395585829fa87b54d3c8c9c"}]},"branch":"refs/heads/master"},"133f4f627c8d06177cf2c783ab89d152ac3cd83b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-08 12:45:28.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/68/870968/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/68/870968/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/3"}}},"commit":{"parents":[{"commit":"8bb44bd2331026aad6b403ac839022f2cf51b1b9","subject":"Return designate tests to admin-plugins job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/8bb44bd2331026aad6b403ac839022f2cf51b1b9"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-18 15:44:41.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-02-08 12:44:18.000000000","tz":60},"subject":"Remove skipped test from designate-tempest-plugin","message":"Remove skipped test from designate-tempest-plugin\n\nWe can remove test_manually_trigger_update_secondary_zone_negative\ntest from exclude_regex as bug 968717 was fixed by removing the\ntest from the plugin [1].\n\n[1] https://bugs.launchpad.net/designate/+bug/1968717\n\nChange-Id: Ie8d0cd289302b789d48c925c1ab0b9940a6e09ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/133f4f627c8d06177cf2c783ab89d152ac3cd83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/133f4f627c8d06177cf2c783ab89d152ac3cd83b"}]},"branch":"refs/heads/master"},"5bc9166bcb459254141292e09389b3ee1becf1c8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-09 11:06:36.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/68/870968/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/68/870968/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/68/870968/4"}}},"commit":{"parents":[{"commit":"7eeb86a34274f3f74e0dbbe55796cf9d10b924a8","subject":"Return designate tests to admin-plugins job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/7eeb86a34274f3f74e0dbbe55796cf9d10b924a8"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-18 15:44:41.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-02-09 11:06:23.000000000","tz":60},"subject":"Remove skipped test from designate-tempest-plugin","message":"Remove skipped test from designate-tempest-plugin\n\nWe can remove test_manually_trigger_update_secondary_zone_negative\ntest from exclude_regex as bug 968717 was fixed by removing the\ntest from the plugin [1].\n\n[1] https://bugs.launchpad.net/designate/+bug/1968717\n\nChange-Id: Ie8d0cd289302b789d48c925c1ab0b9940a6e09ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/5bc9166bcb459254141292e09389b3ee1becf1c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/5bc9166bcb459254141292e09389b3ee1becf1c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[]}
