)]}'
{"id":"openinfra%2Fpython-tempestconf~881551","triplet_id":"openinfra%2Fpython-tempestconf~master~I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99","project":"openinfra/python-tempestconf","branch":"master","topic":"fix/2010573","attention_set":{},"removed_from_attention_set":{"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2023-07-12 13:09:04.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-07-12 11:16:06.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2023-06-28 19:57:29.000000000","reason":"\u003cGERRIT_ACCOUNT_10459\u003e replied on the change","reason_account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}}},"hashtags":[],"change_id":"I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99","subject":"Fix tox\u003c4 issue for zed job","status":"MERGED","created":"2023-04-26 11:02:49.000000000","updated":"2023-07-12 13:11:00.000000000","submitted":"2023-07-12 13:09:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"881551-fix/2010573","meta_rev_id":"b80ba43eca95d98de4ba0066e5fedb796aa58154","_number":881551,"virtual_id_number":881551,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 13:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2023-07-12 11:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2023-06-28 19:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-07-12 11:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 11:38:39.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-04-26 12:19:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-29 16:16:05.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2023-06-06 10:21:20.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"}],"messages":[{"id":"8495f10c4c752634b23433ff0c563d89b84e6785","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-04-26 11:02:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"989cb0b5566304f30e0c7f9c1ecc727952374925","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-26 11:38:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5db94c88ff7717e99ff4de51cd2b0b02542f1f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:19:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30627e33802e499d8b754529679165f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0c11b9349934ba7bf3f5f26ca3b0775 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983c49300049496cad724abb5865ea63 : FAILURE in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6d5edefd5ee44ba9f0f412c4e1a432a : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9e88c50ba1a4d969a786acae099900b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3544567f76d9422fbce227148bbb7294 : SUCCESS in 4m 01s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/2ede4c5864c34a749be2bac4714433b5 : SUCCESS in 47m 40s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/c84918cf19fd4d01b131256066a228f1 : SUCCESS in 1h 01m 38s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/8b48fbf6b12146b7a50aedeab08042a8 : FAILURE in 47m 41s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/154a41b3b04e43a6a1698cadb1eb34c3 : SUCCESS in 36m 26s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/59bcfd05260e44e08aa86883ba41c964 : FAILURE in 50m 40s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/e32111967a224adaaf420b4bc5163b74 : SUCCESS in 45m 36s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/fa7c4de7e8524a4cba2b3ec6b1b96746 : SUCCESS in 46m 53s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/9335504987f0464db4b63278e82b0b83 : FAILURE in 11m 28s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/18deeef5aeed47f087f17f8f00fe552c : FAILURE in 14m 51s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/4c024bf2c0814828b400ba20ec99e7ad : SUCCESS in 1h 14m 17s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/81d1691b250a4640b9a1503685c27744 : SUCCESS in 54m 02s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2e694e9230414da2b486b5c7aa0592b8 : SUCCESS in 3m 55s\n\nWarning:\n  Comments left for invalid file config_tempest/main.py","accounts_in_message":[],"_revision_number":1},{"id":"221a7079aab6e4e04be2c1bfc49c1af97db4a08f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-04-26 13:43:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"776dcad48c9ed29940815ae1c4c4610881a31d56","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-04-26 14:23:31.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"c7e3095d6ea7d9a1628573d6810168b01d2f1b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:01:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ecb273ee026465188bd7311013c6659\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f243496d0fb04dae8bc57a5f5e4a5e84 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef823337bce482f95dcfdf71589b31b : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/566b8f558cc2429aa1181aa5bf3da296 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3607146749b043dc9cad6261cd9e44bd : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023371eeb4b94e53833ede3cfcf5755d : SUCCESS in 3m 26s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/f4dd9fc733424d5e96e09b61e79b6a9a : SUCCESS in 51m 26s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/d3d397b9313b4264bc3f63d2ecd087cb : SUCCESS in 57m 16s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/386ac68fecf341b8b8faa4c3126dcb29 : FAILURE in 29m 10s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/6923c3b828364636a00dd8480478436d : SUCCESS in 36m 09s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/b68aa804aaf64b4599f77aef3cb23f0a : FAILURE in 29m 44s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/f2bd84fa51864218ab457d6bf7dfb74b : SUCCESS in 35m 09s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/e4e465992ea4406995b956fd1b735268 : SUCCESS in 47m 22s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/f3561ecd86fc479b8701ca6e8ef600ee : FAILURE in 11m 29s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/9c8267f4d5674f8c81770ef47bd26094 : FAILURE in 11m 38s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/08e0e66040ad4fc5840740110d7ee82b : SUCCESS in 1h 15m 03s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fa73a23c39184246859f2c8b1c291fd9 : SUCCESS in 55m 19s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1bbf45c403e94adbba85c4740369cada : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"dffac389ab6de34a1ee8aa19afcf5de1e91db8df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-04-26 19:37:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5f247356100b79a5e3fd81f2d0f02745e365767d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 20:54:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a55e761b8c91498d97cc5985eb8008f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b29378122f5242529fb9078025316a8f : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af9bf2851e64560b839e6b213e991b0 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f21cf1d000f4d67b417518c1a6c180f : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e02dc1701ed4699ad06a58643915f46 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/991d0cb5c31c442791a8e2d390c216eb : SUCCESS in 2m 42s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/0a7a12ef03194b039dd107d7f2dd7425 : SUCCESS in 49m 56s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/b7bccadedcde4d78a830d0e450a0838a : SUCCESS in 50m 07s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/37d12b8f47bc4e85b110737ce43c34b1 : SUCCESS in 54m 31s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/c429aaf14f6a41ad8e1760549ee0e216 : SUCCESS in 46m 15s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/793468402b514ace9e9633e4c2de922f : FAILURE in 47m 12s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/d307765618fd40ddbabdc1db7fefa78a : SUCCESS in 50m 28s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/caec9732999f436c9070f00ff5b24b2d : SUCCESS in 50m 35s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/06e10afef487486e94cb828f83e9aaa6 : FAILURE in 12m 23s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/37fca7cae29042aa865d9554030c5e12 : FAILURE in 10m 06s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/62e9bfb431454c98a895637b83e8d7d1 : SUCCESS in 1h 14m 25s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9bc7a4f120aa400b8c81abe877022d91 : SUCCESS in 48m 59s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6993d236df8d496fb1acb17899803919 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7db04cb1fb0c2a0b24669791f37baf810b5bde8d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-09 14:43:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db76663257fa5579d74c8de2ea21dd5e66735a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 16:17:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/922031db3c084b03aa6aac3368217868\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9044ade37014841990429be54058b96 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57b6bdf553a44158ce5b04fa49ffca6 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be482d20156e49deb557d4991071f16c : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cc38a3d7405405e977681f699a9e3b9 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cef01a30d3e430ea7d679f121475b2c : SUCCESS in 2m 56s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/09fbc025f10a42638438665a2cb17507 : FAILURE in 37m 50s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/f60916bb945143f992ac151268fadca1 : FAILURE in 46m 34s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/6e32026e52e84390bdf042585b5e82f3 : FAILURE in 49m 50s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/4d15e1973f324b449968efbad671e548 : SUCCESS in 44m 13s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/a296440fbae94e6c8e98b35f1ade5e8f : FAILURE in 42m 10s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/f84c8af5b0c54413bde6db794b884069 : FAILURE in 57m 20s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/1bfb9434fd1e468687bda5028adfe1d7 : SUCCESS in 35m 20s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/1497527e7fe14066b8087d4f01ba3b00 : FAILURE in 12m 56s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/72629ea58c744ad1b137f50de6564e9c : FAILURE in 11m 14s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/f57eeaa98f0a4aab8a18bfa841f735c6 : SUCCESS in 1h 25m 18s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/db6106afc53c4b188fe546e3555e67ba : SUCCESS in 1h 02m 36s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/594d25b28c044bad9adeb4dcde3b8ebe : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"33e40b3741ed2ddd92c9c38b0f0c0592b7d25ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-05-23 17:23:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8014096a8a71493db3630e5f38d2691993b4736d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 18:54:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77f00442ed7149828b61b9eab53061d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b439d9ca68834344abac35a5f387679d : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b06c8dd2dde4c1a85f5b87bebeed965 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30aae88d4e9e493eb0368d5657761793 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05e3c3a58b7b4f49a506dbc8bf1bac6e : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7af44aa022a447c082fe562e6aad13e8 : SUCCESS in 2m 48s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/64711f8b0e094c2db6fac89c3c97e216 : FAILURE in 46m 53s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/e659be0d13f94bffaa9bd24d7aff50b9 : FAILURE in 37m 42s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/b7c6cf9ce4c04770a67c372e0d6217cd : FAILURE in 52m 11s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/bba941acf8f645358fa05ccb9e38d81e : SUCCESS in 43m 17s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/00a20cbcfb07456f8f298aff287f7458 : SUCCESS in 1h 21m 05s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/ffa3526c0438468989b002bbe172d835 : FAILURE in 47m 25s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/0aed9642e3bd46f788f58e0a6fb94a68 : SUCCESS in 51m 09s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/20b4302ec79d449dbb42abff054e9313 : FAILURE in 20m 04s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/9a0242e2450f420197e0b037b1b44176 : FAILURE in 11m 35s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/d24fb2ff9c0f4a7a9309bca51f9587be : SUCCESS in 1h 24m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3a4e5f9a26374e4581c8b173a019d195 : SUCCESS in 1h 15m 34s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d4bf90e3daf347f0a34da9f8e787bd68 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"99f2827df2c15d972a46695080c2470b6fcba6b0","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-05-23 18:59:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b1e61cc4e2b6ff25c84964bc9d0d98004b414f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 20:32:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09f85379dbf24351a1d0cbab3bb3fd7a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9054c930367744c587d9545e262c9475 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521febbf40b345e2ba0f07219ef5f88b : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c13f220afb4841328cefa426e0df1067 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3823039d0e44553a690f04a0f125337 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ddc60f3e9b245d6acecdfd48891cbf7 : SUCCESS in 2m 32s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/d8a33c64e750462e845422b1d2a2d523 : FAILURE in 46m 27s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/b546a0b0e8b84a0e9b73e94c34fadb82 : FAILURE in 49m 47s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/f33ad54251c747f1bed5019e9fec3a6a : FAILURE in 56m 52s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/7db9860831fd486ea239ad448f526ef6 : SUCCESS in 49m 23s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/fe501a7eb9f14064b580f394ea6a606e : SUCCESS in 1h 26m 42s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/9f6243ff392c48ecaaf20261ca09515f : FAILURE in 57m 58s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/48d55a36a2a141ff9e874db561306db5 : SUCCESS in 55m 23s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/3071ba8da1874313996f0b1979b8e821 : FAILURE in 20m 49s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/e4ec8b0ec7134d17a9e22a32394b0b91 : FAILURE in 14m 10s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/dc0a2e7ce16e4ca394523965dfd84bbe : SUCCESS in 1h 14m 30s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d2a7629a624642b7ab6139c3e1c27488 : SUCCESS in 58m 39s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/51ae439b963746059414d3b07747b0d3 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":4},{"id":"f825a61ac839d4e44a60e675966d0e09780e43d8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-24 07:53:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2d2baba30da93ad241d29eede033f758928383db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:27:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ff8c1c88409440f9cd43f27911bb9c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/352c6d94471048d897a40d3e7eadcd0f : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21403422c4a143d8b00e53043c43c48b : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a7adef8880445c783020680d997e7d2 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31960873bfea495882cf299fa6a0f82c : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f015a945304b4088af5bcd5fea71421f : SUCCESS in 3m 16s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/d726ea56a6374ef995e8828fdef8c739 : FAILURE in 52m 48s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/f2e2c1da88a44a9f99f3a02adc4388e1 : FAILURE in 58m 08s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/b8a90c8f90f6449eac2c11aff1de2c61 : FAILURE in 43m 11s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/31113fe977e047d0bb721769648de9b9 : SUCCESS in 49m 39s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/158d9ea61bb3484890285f570106413e : SUCCESS in 1h 15m 34s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/987e1c9baf0046c5ba8b90b9548d9eea : FAILURE in 39m 22s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/ddd04f8ba8ff4b77893c1a60e16b4d0e : SUCCESS in 34m 12s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/d7802021c4344c69b975d5e04b03786e : FAILURE in 11m 17s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/40eb2e592b154c41979d4b54d0a0a67e : FAILURE in 25m 08s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/2704a561948247d88741a1b4e4b4b5ff : SUCCESS in 1h 25m 25s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c68b62272cec4c9baa10a2d4e58d5d69 : SUCCESS in 1h 16m 48s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a7a36b818ccd440c9b4300bad419f0f6 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a9f4aab102d16457b52befaa87e7dfbdb8c15454","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-24 11:10:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"17b5f8c9e0881d43f9cea47bfa103f63826018ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-29 15:18:49.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4949536fd69b9e5fddc389581858a6f9f5ca30f6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-29 16:16:05.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"96ce8a4200ab7a56600c3dfe7e0ab5b4f21f66e8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-29 16:23:06.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"72364fb0a1551dc658d21dc4f00efbc903d1dcf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 17:05:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e85db9aef5e4f20b642b18f58d91a94\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc06d4d7c55949078846d91e6d633baa : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1e494b9eb1433795c91c902679a8c6 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b20f93c47e16479f80b055a13c08ef6c : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00e3acf580194abcaf88941e36cb5f2e : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9783169e0b34a39b5d6678202ff2c17 : SUCCESS in 3m 13s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/0f5686dc7fa2498181de7523c10e7809 : SUCCESS in 53m 13s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/87128dc63c2642fabd47ffe1ba172eaf : SUCCESS in 50m 41s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/35a86e42503e4173a7c2d731c2d3f26d : SUCCESS in 42m 26s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/903ee956d1f54a04ab8dcd117929345b : SUCCESS in 48m 53s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/4c0b07b5098041bcbd4d4749269e546b : SUCCESS in 1h 22m 26s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/6ca0162de857452d91830411e04b50f9 : SUCCESS in 53m 17s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/276afa2d4e2140d693bbb7bb61c2695e : SUCCESS in 46m 34s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/f23acb5b8c464d6da5460e38c76ab2ac : FAILURE in 12m 57s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/a9ea971a8e15422b957120c7d01951b4 : FAILURE in 22m 47s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/289286fce5d046589160cd08a53320d3 : SUCCESS in 1h 37m 06s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/23e0c5e7e5174fedba7221b2e617de43 : SUCCESS in 1h 21m 26s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4c29b2b0590c4fe288ac3dee41587b91 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"1620909f0e4b05951547d4dda35f1ef11ebf62c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-06 09:14:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1bc123826fe8e2ae5c4a60488210c54ec098beff","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-06-06 10:21:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7702386db146564cc82a6c372c7690711f4b8460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 10:43:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042e59bb72674c4b8b4425010262f9c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64c898b6a4ab4733bdb313141aea06dd : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6950091cd28541699f7673356f29aa1a : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8928297363f4d238b4052663369f790 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25a2c28c26d641d0afea80d15fc57ff2 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64406b62db904424add205e16057d96a : SUCCESS in 3m 01s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/f38ce1874e88442c8c3daa94d9019a70 : SUCCESS in 41m 26s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/e2850ebfb0d24cd9851bf42d0d26bc4f : SUCCESS in 35m 59s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/c117f5766ffe402f846073ddb49892ac : SUCCESS in 42m 52s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/265b458cbc7e4af4b515bfbe1899cd39 : SUCCESS in 37m 33s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/f84cfc3a6ba1442c877dbc7527064a78 : SUCCESS in 1h 07m 33s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/8c34e5d56cd94cf7a91e233de8011720 : SUCCESS in 40m 46s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/b10194919e5f4819849b590dd131ebc6 : SUCCESS in 51m 38s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/2e3a45973b204c49ae818dc7794fd86c : SUCCESS in 37m 04s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/0fb4ab3f57114e129454f0b9cb8a0ba3 : SUCCESS in 30m 23s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/ef509e8f64514fea922680d6db8288f9 : SUCCESS in 1h 26m 12s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c5f2fd33760b48378d02f126f7923506 : SUCCESS in 1h 04m 14s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/43a1ddbc33fb472284061b03528ef249 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"92e8fa2494d8fa8e4c7814cfebeeec4df1c830fb","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-28 14:29:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"90bb00c351460438875afded1c31bb159a5f695c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-06-28 14:43:35.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2f1fdc555b3a96749c2c801d5e18008e16551920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-28 17:05:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d595d05f9ea8b2a992cd0854f8b32382bb2a1bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 18:52:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a53e39262945959f4b30e0ea9bc92b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0a6f8a8ffa04a749a902e2d7cc18ec4 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa27459b5abc4e52b33479c2e7fe7df8 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ac48cba003f444cb1a0000b07708753 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8619c57c3df94e4fb111e3ae01100359 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/871e02e907a74141a799feaf2e274f27 : SUCCESS in 4m 52s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/675e1eb8091449d793298b8f79f18bb7 : SUCCESS in 1h 07m 12s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/b4830f326d494e968cb8ac4315eeaa7f : SUCCESS in 1h 08m 07s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/1747e2a707674e5882c19ad66ee30355 : SUCCESS in 1h 01m 19s\n- python-tempestconf-tempest-devstack-admin-2023-1 https://zuul.opendev.org/t/openstack/build/332c59b283cb4475bf0bc0f9c0b58d82 : SUCCESS in 1h 04m 41s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/eb78e301abc24add9d919c6f73a22a87 : SUCCESS in 59m 28s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/d8b53ed04c944b1fad58d4495c154d9d : SUCCESS in 1h 36m 27s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/2f40f662ec00478bbb190ec12baea9e2 : SUCCESS in 1h 07m 18s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/1d089a9c6c854611a2316e9f90d764c3 : SUCCESS in 38m 19s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/c975ab9140b2468f943319536c54e6ae : SUCCESS in 32m 32s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/c40630c52bbd494aae87371d40a581a7 : SUCCESS in 57m 42s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/a3dd6ee5d85046b4b4fc62b4d3158181 : SUCCESS in 1h 32m 34s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0087eabccea8425c98c0dec402eb511c : FAILURE in 28m 41s (non-voting)\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3a755519296b4c7b8ea39a364bec1bbe : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":7},{"id":"458ba943b512767082c931ce38797b32fa735ebb","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-06-28 19:57:29.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"fd35d624594ab9d00b761b8af0c438abddc9f39b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-12 11:16:06.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e5f88d953c985c7ef6092f24ae6f9644582e88aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 11:16:52.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ecea70c19f6e06a21c9b2b8802a4cf97e8dd8fb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 13:09:04.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5bc1f01312948ebb10482b25cfda9a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4df95ab049448a87f99d49b7ca652e : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/176933cb6e2a4b259bbc387fc74fd295 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/339bd53cf4b640198336858d398756bf : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88f899345e384b94a3efbd225e580bfa : SUCCESS in 3m 01s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/1b6ec881a6d44e34a4622db60d6079e8 : SUCCESS in 53m 00s\n- python-tempestconf-tempest-devstack-admin-yoga https://zuul.opendev.org/t/openstack/build/715e19a8fec048e58000e76f05298985 : SUCCESS in 1h 00m 25s\n- python-tempestconf-tempest-devstack-admin-zed https://zuul.opendev.org/t/openstack/build/12a5360af6bc406884a52940555b0092 : SUCCESS in 1h 05m 49s\n- python-tempestconf-tempest-devstack-admin-2023-1 https://zuul.opendev.org/t/openstack/build/708fe67d0632421c858c41486daf57e1 : SUCCESS in 1h 10m 55s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/8d46e9743dd04fa3b1a68b98523c2aa7 : SUCCESS in 1h 00m 40s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/1b718cc3f2b748c09bf5612fd5d72568 : SUCCESS in 1h 26m 19s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/1ba26ea7db4d4aa2adb71bf801b01949 : SUCCESS in 1h 41m 42s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/94ac66d7eb3a4fb99e39393c731bda63 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":7},{"id":"930cb7c6ed79312cb5b015a1db092b37225fb5fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 13:09:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b80ba43eca95d98de4ba0066e5fedb796aa58154","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 13:11:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/927cc66708b64bebb7c6a0680cff2da1\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/be58edd990d349faa93e758251d3cf2b : SUCCESS in 36s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9ff70dd6bc8bbddaed80e909084641f3ce941456","revisions":{"8ab9d32c4f94c6cdbfd0c89a57d62da33e622dbd":{"kind":"REWORK","_number":1,"created":"2023-04-26 11:02:49.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/881551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/1"}}},"commit":{"parents":[{"commit":"b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d","subject":"Releasenotes for 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 11:02:37.000000000","tz":120},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again\n\nStory: 2010573\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/8ab9d32c4f94c6cdbfd0c89a57d62da33e622dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/8ab9d32c4f94c6cdbfd0c89a57d62da33e622dbd"}]},"branch":"refs/heads/master"},"faa6d77d04c497d41a736c79ee15530f8751e228":{"kind":"REWORK","_number":2,"created":"2023-04-26 13:43:07.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/881551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/2"}}},"commit":{"parents":[{"commit":"b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d","subject":"Releasenotes for 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 13:42:54.000000000","tz":120},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again\n\nStory: 2010573\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/faa6d77d04c497d41a736c79ee15530f8751e228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/faa6d77d04c497d41a736c79ee15530f8751e228"}]},"branch":"refs/heads/master"},"943dba4b971db1dff974379df6a211b6ab1b9ba6":{"kind":"REWORK","_number":3,"created":"2023-04-26 19:37:03.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/881551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/3"}}},"commit":{"parents":[{"commit":"b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d","subject":"Releasenotes for 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 19:06:56.000000000","tz":120},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again\n\nStory: 2010573\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/943dba4b971db1dff974379df6a211b6ab1b9ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/943dba4b971db1dff974379df6a211b6ab1b9ba6"}]},"branch":"refs/heads/master"},"ffbfafe9c90d7e2af576728c1cf6204f172df5c1":{"kind":"REWORK","_number":4,"created":"2023-05-23 17:23:58.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/881551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/4"}}},"commit":{"parents":[{"commit":"b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d","subject":"Releasenotes for 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-05-23 17:20:36.000000000","tz":120},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again. Set\nensure_tox_version to \u003c4 for zed admin\njob and admin plugins job (since it failed,\non tox\u003e4 related issue too.\n\nStory: 2010573\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/ffbfafe9c90d7e2af576728c1cf6204f172df5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/ffbfafe9c90d7e2af576728c1cf6204f172df5c1"}]},"branch":"refs/heads/master"},"96ba08311b79a3b44c6e4631d338890a7c7970df":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-05-29 15:18:49.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/51/881551/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/5"}}},"commit":{"parents":[{"commit":"b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d","subject":"Releasenotes for 3.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/b0ee1bcdacd85b073b77ad338d3d78aa16a3ef3d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-05-29 15:18:49.000000000","tz":0},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again. Set\nensure_tox_version to \u003c4 for zed admin\njob and admin plugins job (since it failed,\non tox\u003e4 related issue too.\n\nStory: 2010573\nTask: 47319\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/884584\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/96ba08311b79a3b44c6e4631d338890a7c7970df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/96ba08311b79a3b44c6e4631d338890a7c7970df"}]},"branch":"refs/heads/master","description":"Edit commit message"},"126b4c011e72cfafbde7795e0fd9b974fe8a7fd6":{"kind":"REWORK","_number":6,"created":"2023-06-06 09:14:41.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/881551/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/6"}}},"commit":{"parents":[{"commit":"ea9dc5727813ca363b3965d69124b374ef9b9932","subject":"Handle when there is no volume backend available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/ea9dc5727813ca363b3965d69124b374ef9b9932"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-06-06 09:14:28.000000000","tz":120},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again. Set\nensure_tox_version to \u003c4 for zed admin\njob and admin plugins job (since it failed,\non tox\u003e4 related issue too.\n\nStory: 2010573\nTask: 47319\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/884584\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/126b4c011e72cfafbde7795e0fd9b974fe8a7fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/126b4c011e72cfafbde7795e0fd9b974fe8a7fd6"}]},"branch":"refs/heads/master"},"9ff70dd6bc8bbddaed80e909084641f3ce941456":{"kind":"REWORK","_number":7,"created":"2023-06-28 17:05:47.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/881551/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/51/881551/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/51/881551/7"}}},"commit":{"parents":[{"commit":"468044fa9899b8cd97299f4702251ec4db9eaa55","subject":"Switch to bobcat testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/468044fa9899b8cd97299f4702251ec4db9eaa55"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-26 10:57:08.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-06-28 17:05:38.000000000","tz":120},"subject":"Fix tox\u003c4 issue for zed job","message":"Fix tox\u003c4 issue for zed job\n\nAdded task to install proper version\nof tox(\u003c4) so the job won\u0027t fail and\nset the job as voting again. Set\nensure_tox_version to \u003c4 for zed admin\njob and admin plugins job (since it failed,\non tox\u003e4 related issue too.\n\nStory: 2010573\nTask: 47319\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/884584\nChange-Id: I0e9ad3632d16a5aa8c0e98d3e0d10bee79cead99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/9ff70dd6bc8bbddaed80e909084641f3ce941456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/9ff70dd6bc8bbddaed80e909084641f3ce941456"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
