)]}'
{"id":"openinfra%2Fpython-tempestconf~918038","triplet_id":"openinfra%2Fpython-tempestconf~master~Ib10cae1fa3610c4ea0a54b7f4ebbdb146bf4fce2","project":"openinfra/python-tempestconf","branch":"master","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2024-05-06 13:27:23.000000000","reason":"Change was submitted"},"36227":{"account":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"last_update":"2024-05-06 13:27:23.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-05-06 08:01:01.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-05-06 07:59:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2024-05-06 13:27:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib10cae1fa3610c4ea0a54b7f4ebbdb146bf4fce2","subject":"Prioritize external network named public","status":"MERGED","created":"2024-05-03 11:09:29.000000000","updated":"2024-05-06 13:28:19.000000000","submitted":"2024-05-06 13:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"918038","meta_rev_id":"f28eb22b64b1f2db4bb95653444727192421b787","_number":918038,"virtual_id_number":918038,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-06 13:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2024-05-06 07:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":1,"date":"2024-05-06 08:01:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-05-06 12:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-05-06 12:02:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 11:10:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2024-05-03 11:10:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"state":"REVIEWER"},{"updated":"2024-05-03 11:10:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2024-05-03 12:42:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-05-03 13:09:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 08:01:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"02453729ff5b4fe3b92699e3528cdf7d76ed4fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-03 11:09:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a0a188cf4ebcb0c01aa6ebc0a2a3e45ada7a26a","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-05-03 11:46:35.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fd2f0b56619fe31e7e48cd3183f9339521939660","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-03 12:42:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc428295913adea6ef05999103b03389153193f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 13:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcbceea567b24eaf9d4d0e37449634dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08adfdbb75474641886294c59034f9c7 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ff9541b381b446dae1084fe804a170b : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/447427341ae142e1bc235a486fcdc1c7 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68b012fa94eb4fbe97a58a31ef43b391 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1f2e5119c484116bbda0f2519bfc84d : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db2415ccef324fe2b674bafa2b4db20a : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ee922ee72694c9eac34fe6471f935f3 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9acab51f875a4cef94297f5896340126 : SUCCESS in 6m 29s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/0eeb5b5bb3674e6fb024f72bf4d2a16b : SUCCESS in 55m 45s\n- python-tempestconf-tempest-devstack-admin-2023-1 https://zuul.opendev.org/t/openstack/build/d9c2a557feab4cccb13c8750cb96f07a : SUCCESS in 36m 48s\n- python-tempestconf-tempest-devstack-admin-2023-2 https://zuul.opendev.org/t/openstack/build/eda6c377238d4eeb93a15c3ef6b2cd2d : SUCCESS in 54m 18s\n- python-tempestconf-tempest-devstack-admin-2024-1 https://zuul.opendev.org/t/openstack/build/e0444eb8413b4178aa99718c67736c84 : SUCCESS in 1h 00m 35s\n- python-tempestconf-tempest-devstack-admin-fips-centos9 https://zuul.opendev.org/t/openstack/build/161dc1eff8234812ab7a6e76206f90d7 : SUCCESS in 56m 50s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/18c83afb82f74eb1ba8e86ce2d29efd0 : SUCCESS in 49m 52s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/98984553afbc4eb698344f4a1932e44b : SUCCESS in 1h 11m 57s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/65d734914a3743c298ac21c952824d31 : SUCCESS in 49m 17s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/e8689754d6874ebf8ffa68bb0bb87359 : SUCCESS in 42m 58s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/4b5a08ac2bfe488c9bed4c4084d6c1bc : SUCCESS in 35m 15s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/01ba7e78757149bc8871d5f5f33eb213 : SUCCESS in 32m 44s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/06f77cbea0504847bce8e2bcce979789 : SUCCESS in 1h 23m 17s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/be43d49dbac746c3b658db2fafcad0df : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e6641c85d909ad8212f933ed375029020b0e89dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-04 17:06:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2dde821e092097556a1b04589d78c71693b96ff","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-04 17:06:35.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e653d520c6352d057e1f6887aef52dc710af7298","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-04 17:06:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ef84e2be4accf2335f2d69cc2d32d6ac46720a1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 18:33:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad788fca3ee849a2bd5ea7c89788afec\n\n- python-tempestconf-tempest-devstack-demo-2023-1 https://zuul.opendev.org/t/openstack/build/424330503efc463f970ac201e85eedd7 : SUCCESS in 47m 22s\n- python-tempestconf-tempest-devstack-demo-2023-2 https://zuul.opendev.org/t/openstack/build/88ccdf5d856f45bab5386799f9bb3392 : SUCCESS in 49m 10s\n- python-tempestconf-tempest-devstack-demo-2024-1 https://zuul.opendev.org/t/openstack/build/8b227872b842499b99b107b94bb628c5 : SUCCESS in 36m 18s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/53bdae37ebc94c41bbe1304f192e378b : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":2},{"id":"90ab4c0ce1474dbb3fd6d1595b897e8fc5808c11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 18:42:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b813cd57dc514e6c806cef52e4ce6f99\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2970b2bedbce45329f65cd2419990e53 : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f847a9bd974427e9439d401919e5463 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae68a0612fc248f6912a50e70c5f0dc8 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7323e57f094d4c6e8ad5b3ff2cc38809 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/149ae575c9f142f5b86460fba5ecef2e : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f22bb80d5d74a9bb2c90576598b6e32 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63e73199d8e147418aae15bf077db35a : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8e6e25bfa0e42ea8d4f66c5270ce67f : SUCCESS in 8m 14s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/9a7e5b24d2664bfb919911a47c9356a2 : SUCCESS in 43m 10s\n- python-tempestconf-tempest-devstack-admin-2023-1 https://zuul.opendev.org/t/openstack/build/7f9832ed4fa447cd9990453348e54ee6 : SUCCESS in 46m 14s\n- python-tempestconf-tempest-devstack-admin-2023-2 https://zuul.opendev.org/t/openstack/build/afd0dea9730d48fabb0abdf809521f93 : SUCCESS in 44m 40s\n- python-tempestconf-tempest-devstack-admin-2024-1 https://zuul.opendev.org/t/openstack/build/e13b27ffb6094a1ab74c76bc7e334912 : SUCCESS in 48m 13s\n- python-tempestconf-tempest-devstack-admin-fips-centos9 https://zuul.opendev.org/t/openstack/build/c338ac749cef4a2cad3b58eedb95f17d : SUCCESS in 1h 04m 02s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/9ebfb19565c648d4b4c75d6b6cebeef8 : SUCCESS in 41m 04s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/ddd21a0fbbec46c3a2690546920742e3 : SUCCESS in 59m 10s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/dd6bd039792142228d886aa1f6623fd6 : SUCCESS in 40m 34s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/bb70efe1a59e4676a9e749d834fb07e0 : SUCCESS in 31m 55s\n- python-tempestconf-tempest-packstack-admin https://zuul.opendev.org/t/openstack/build/9792baecae714feab8d39cc747c15d5a : SUCCESS in 28m 35s (non-voting)\n- python-tempestconf-tempest-packstack-demo https://zuul.opendev.org/t/openstack/build/03384542beb34f55825a51af47b8240d : SUCCESS in 34m 23s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/3961e38724ee43408a62ef40c16dc0c1 : SUCCESS in 1h 26m 22s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/380d177220a141aa8d347c65b7c550d6 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"ed7900e1f3aba7e0b8417a79e188bc851621f8c7","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2024-05-06 07:51:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"67685a670dbf9941170d367cebc18adf7edfa456","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-06 07:59:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3630e21ea3ad1ba6ee26beae20a2f16985734bc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-06 08:01:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f5556d48aba55eaea10f79533dc2a1e3aebf425","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-05-06 12:02:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bca2d938eef84ee42935c9d6be1298119c9e40da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 12:03:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f79af6e01ac495f12d8c1402777fc6283d08105b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 13:27:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77dd1fbc95f4ffebfcc691dc6418f44\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc30f124ec0c4ee6b0f33200e41a3a55 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b67a2f408844f9876808e584032c54 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3272edffe3964dde80178591127ede48 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be69635b3f29437c8fb3d8f1d11840da : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff9774dda5754ce386fe6b87adc965e7 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1110768a53364a6eaaf135c8efec8805 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/564e8530c6a24134b444cf6945363631 : SUCCESS in 5m 32s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/49064cc6f7b24dedb5aa8a3cdfcfe2ec : SUCCESS in 43m 44s\n- python-tempestconf-tempest-devstack-admin-2023-1 https://zuul.opendev.org/t/openstack/build/26de0b5e60f441e282b1e72dd9b11073 : SUCCESS in 40m 24s\n- python-tempestconf-tempest-devstack-admin-2023-2 https://zuul.opendev.org/t/openstack/build/5ede6360360245b0ac9893dc49758894 : SUCCESS in 44m 08s\n- python-tempestconf-tempest-devstack-admin-2024-1 https://zuul.opendev.org/t/openstack/build/4b1021d2abfd473897025b0ffaeb6e6a : SUCCESS in 37m 07s\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/520ce49e3fcc4e54aac063a40d59826a : SUCCESS in 47m 26s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/80d11714e8ba444d9c7481818825853a : SUCCESS in 1h 22m 41s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/e5a7737984994215a7723b8a0f374d0f : SUCCESS in 1h 13m 42s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e8dba29ba9ac4bd881804350ade27088 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c1c192eef66536c46719bffbccd0d84fb4d3a6dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 13:27:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f28eb22b64b1f2db4bb95653444727192421b787","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 13:28:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f14c5686a8a44aebe7f98139abf9ef5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70e0151ac96f4835b50b1bb0870a7c34 : SUCCESS in 40s\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/bb86fe5761764999ae252fe838295cab : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4d8dc124ca8abb0f2aea7fb9fdc228e72354d0a","revisions":{"38b03d26450c4b3c1aef56bb84fb10dd1c887f93":{"kind":"REWORK","_number":1,"created":"2024-05-03 11:09:29.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/38/918038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/38/918038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/1"}}},"commit":{"parents":[{"commit":"46fb09a653bd89b2deae6c25836ba2234409439d","subject":"Merge \"Update jobs for the current cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/46fb09a653bd89b2deae6c25836ba2234409439d"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-03 11:05:54.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-03 11:08:26.000000000","tz":120},"subject":"Prioritize external network named public","message":"Prioritize external network named public\n\npython-tempestconf will prioritize the external network named public\nwhen auto-discovering an external network for tempest.\n\nThe public network is the network we end up using in most of the\ncases. This will help avoid potential conflicts with other available\nexternal networks that are not meant to be used by tempest.\n\nChange-Id: Ib10cae1fa3610c4ea0a54b7f4ebbdb146bf4fce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/38b03d26450c4b3c1aef56bb84fb10dd1c887f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/38b03d26450c4b3c1aef56bb84fb10dd1c887f93"}]},"branch":"refs/heads/master"},"e4d8dc124ca8abb0f2aea7fb9fdc228e72354d0a":{"kind":"REWORK","_number":2,"created":"2024-05-04 17:06:28.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/38/918038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/38/918038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/38/918038/2"}}},"commit":{"parents":[{"commit":"46fb09a653bd89b2deae6c25836ba2234409439d","subject":"Merge \"Update jobs for the current cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/46fb09a653bd89b2deae6c25836ba2234409439d"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-03 11:05:54.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-05-03 12:40:40.000000000","tz":120},"subject":"Prioritize external network named public","message":"Prioritize external network named public\n\npython-tempestconf will prioritize the external network named public\nwhen auto-discovering an external network for tempest.\n\nThe public network is the network we end up using in most of the\ncases. This will help avoid potential conflicts with other available\nexternal networks that are not meant to be used by tempest. If no\nexternal network with name public is found, the network discovered\nlast will be used.\n\nChange-Id: Ib10cae1fa3610c4ea0a54b7f4ebbdb146bf4fce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e4d8dc124ca8abb0f2aea7fb9fdc228e72354d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/e4d8dc124ca8abb0f2aea7fb9fdc228e72354d0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
