)]}'
{"id":"openinfra%2Fpython-tempestconf~996940","triplet_id":"openinfra%2Fpython-tempestconf~master~I5194ea3d4a431710ef293250a2ae947965e2c509","project":"openinfra/python-tempestconf","branch":"master","topic":"glance-s3-store","attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-07-15 14:32:39.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},"36227":{"account":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"last_update":"2026-10-05 05:42:19.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2026-10-05 05:42:19.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-10-06 10:31:59.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5194ea3d4a431710ef293250a2ae947965e2c509","subject":"Add Glance S3 store backend support","status":"NEW","created":"2026-07-12 12:55:05.000000000","updated":"2026-10-06 10:31:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6fc7a4c2c2f5774df429c4e84cf7828c5fd028d2","_number":996940,"virtual_id_number":996940,"owner":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-05 06:40:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-12 13:45:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-15 14:32:39.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-07-15 14:32:39.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-07-15 14:32:39.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2026-09-02 12:44:21.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"state":"REVIEWER"}],"messages":[{"id":"ecd1fc530d76e7767e73db293d41d165a1f3373c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-12 12:55:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70011cd6fb4ca9bca4578eca51e1dd9a9eddeac2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-12 13:03:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"bca9ad400d6ca12d6782774e7d9dc7a3ad406be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 13:45:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cbabdcdb10c426a983fda2bf23ee8b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/404d47a99da548abaf90f6a7af7e5f10 : SUCCESS in 2m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a19ea2ee1e6b475392b8372ab38ea08b : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff82926c24b742f7b3fc61190f8856a1 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c2a82ed71ec48e78fb57a461c51d98a : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05b6c312c8344a8d94e38ae424dccfbb : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e2f5958efb549e6b65998c222e2f002 : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e6b9ae2eef548aba851a63ec44607de : SUCCESS in 7m 09s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/c722763b0be24ab8a0bcb0682dd5b1d2 : SUCCESS in 22m 24s\n- python-tempestconf-tempest-devstack-admin-2025-1 https://zuul.opendev.org/t/openstack/build/6eb7e6864e8749bea2cbaf338de0e7b9 : SUCCESS in 25m 53s\n- python-tempestconf-tempest-devstack-admin-2025-2 https://zuul.opendev.org/t/openstack/build/c7d4202cfee54f0caaa43955a874a20d : SUCCESS in 31m 37s\n- python-tempestconf-tempest-devstack-admin-2026-1 https://zuul.opendev.org/t/openstack/build/b6fcd78283da41a0a7476e5262b115d4 : SUCCESS in 35m 41s\n- python-tempestconf-tempest-devstack-admin-fips-centos9 https://zuul.opendev.org/t/openstack/build/db59b432f4284640a7d3f683d0a35327 : FAILURE in 20m 16s (non-voting)\n- python-tempestconf-tempest-devstack-admin-rocky https://zuul.opendev.org/t/openstack/build/f35f75d67f9941ee983cd6c9e313e92d : FAILURE in 14m 48s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/5d137241da6445319f134dd33935d025 : SUCCESS in 24m 28s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/fca79edb153b4c6085c25a5962dd6b1f : FAILURE in 24m 09s (non-voting)\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/6b654270204a406aa28520b2ceb2e4ff : SUCCESS in 44m 44s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/11e14e9741174a5e9804b80da818cb23 : SUCCESS in 24m 51s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/55a43de6bbb74acba09e7c317aee4df1 : FAILURE in 1m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"a87be4dfd38a03d9add72dfa6b54ca9032e6cf09","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-12 15:11:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"75ab5499b1cf1a3373e16fa77b553b9869a9d17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 16:05:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80df74407f4048899d29e8755ac1cb88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/333aa9829f2c4abc8ac3b58f58fd901c : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bef16f9ab884f8fb8c50cd94e78448a : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1128fe7974764b43a10a4708a51c783a : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cda4424c8bd40319cdfc59b6ce63847 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b57d75356d34063bb3b7563162e9317 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10fad782f577487a8bf6ec166914e7e8 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88bf5ef05f984a129a39721096ce0e8a : SUCCESS in 5m 13s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/72196ff605f745b381611828becfdc9d : SUCCESS in 46m 53s\n- python-tempestconf-tempest-devstack-admin-2025-1 https://zuul.opendev.org/t/openstack/build/8094e93655f04c65b62a026b7d61dae0 : SUCCESS in 47m 57s\n- python-tempestconf-tempest-devstack-admin-2025-2 https://zuul.opendev.org/t/openstack/build/057f8fd4d69b4d9e875169660dbff55e : SUCCESS in 46m 12s\n- python-tempestconf-tempest-devstack-admin-2026-1 https://zuul.opendev.org/t/openstack/build/c554b590a95a48a3b187d91565a15f1f : SUCCESS in 49m 08s\n- python-tempestconf-tempest-devstack-admin-fips-centos9 https://zuul.opendev.org/t/openstack/build/97c7241110c94e3fa42cff5b3c5f156c : FAILURE in 19m 01s (non-voting)\n- python-tempestconf-tempest-devstack-admin-rocky https://zuul.opendev.org/t/openstack/build/63ca3fc8e63948828d4118cae8aed6ca : FAILURE in 15m 36s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/7f7c8268aa0e43858d9a65a556c17389 : SUCCESS in 37m 21s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/b7c4cd5faf1241f083e8126d28741e2d : FAILURE in 23m 06s (non-voting)\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/34a131ab38b3494882c0f195bb95645b : SUCCESS in 49m 19s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/dfbfdd34a5e040249a506685da0d0a76 : SUCCESS in 17m 11s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/cfc2b4943b5e4734846a19d3fb0e717a : FAILURE in 2m 33s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"b61cb2e76e4caf06ab13c5f298ea934daaab47fe","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 14:32:39.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"12cd7eb531d5b4bcec103a453c8435f22ffbf69a","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 14:45:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f30ca17f9586e83e9a3b5ee8ceeb19364a8eb3cc","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-26 07:56:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"660f3687143a62b25115867c0e55b3fa767e9024","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2026-09-02 12:44:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e27c6f43b88067be3a746b6ed1903288c7ff96f0","author":{"_account_id":36227,"name":"Katarina Strenkova","email":"kstrenko@redhat.com","username":"kstrenko"},"date":"2026-09-02 14:04:37.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b79d0cf9500af91f9b3a5f76b4022907ba9bb11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-10-05 05:39:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1ec17417318580868481d6a9c62559d442ee1160","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-10-05 05:42:19.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9a3f4f25e5bcc9313528c257f677f74e7ffa07d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 06:40:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73c3b78c39ab4d79813b722e803e7b18\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43e94f8d7df944c68c75a04609db65ed : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39df2edcaa114450bb8d66e08419834e : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72df8cc1655406eb9c692d127485c2a : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d736325ab694b64bcbb129eb5230279 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58bc8d7e9ffc41219fe61cc8a5e433ff : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e006fb852c594f55822afcd6612c04fd : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d110c9c319004e70bcf9f06acff6896c : SUCCESS in 5m 31s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/03034cdc30a040ac8cfcf4230641b338 : SUCCESS in 50m 55s\n- python-tempestconf-tempest-devstack-admin-2025-1 https://zuul.opendev.org/t/openstack/build/981f872ba74c4cc19ed1375d3f7e8e92 : SUCCESS in 51m 43s\n- python-tempestconf-tempest-devstack-admin-2025-2 https://zuul.opendev.org/t/openstack/build/3af65a30a6b54fe789017c4ffd3280ad : SUCCESS in 34m 05s\n- python-tempestconf-tempest-devstack-admin-2026-1 https://zuul.opendev.org/t/openstack/build/9ff5c44fcfca4cbe8118aa7d35f7f831 : SUCCESS in 47m 21s\n- python-tempestconf-tempest-devstack-admin-fips-centos9 https://zuul.opendev.org/t/openstack/build/6fead754d8ad4f0eb10b3625a85de4aa : FAILURE in 8m 27s (non-voting)\n- python-tempestconf-tempest-devstack-admin-rocky https://zuul.opendev.org/t/openstack/build/53b310d33717444ca4f8709802320426 : FAILURE in 7m 12s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/1ba4546a141b4178b04f2d086d0e97b2 : SUCCESS in 20m 03s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/0f3b121f653a434bbf71b916b0fc985d : FAILURE in 25m 25s (non-voting)\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/5352aab488d54c70b031cccbd6f8fd3d : SUCCESS in 20m 26s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/5dcd72e6baa244018d835969746deab3 : SUCCESS in 17m 27s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7589ee0917504af5bc73e38b5bdb48e5 : FAILURE in 2m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"5e7f218d6f75e06d458705e653311e796797dec2","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-10-06 09:30:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6fc7a4c2c2f5774df429c4e84cf7828c5fd028d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 10:31:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b389c89a97a540948b887438aa6aacc8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc8fa67e82bb4be9b30e296fef1e6121 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/283ed25cbccd4784bf96c7ce4c05b55c : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716a750fcc464d30809a1ab1b26e1dd0 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a728d15d8d6240fe862e410ff82a5a50 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2af74e1f7f27407081c2ef53386d2971 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1404bf362f84363ac9a42bab93954ef : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96d9ca8ce7e94f52829805c0c23ecc50 : SUCCESS in 8m 49s\n- python-tempestconf-tempest-devstack-admin https://zuul.opendev.org/t/openstack/build/fa86f6ee9085421493c56f3ec42bc1a8 : SUCCESS in 49m 02s\n- python-tempestconf-tempest-devstack-admin-2025-1 https://zuul.opendev.org/t/openstack/build/212b2062f87c4549a755bbb6cc2dd8d0 : SUCCESS in 44m 33s\n- python-tempestconf-tempest-devstack-admin-2025-2 https://zuul.opendev.org/t/openstack/build/6b80c37ed9af4b2aa907ce92e5481f18 : SUCCESS in 41m 14s\n- python-tempestconf-tempest-devstack-admin-2026-1 https://zuul.opendev.org/t/openstack/build/353224ca2a8a4a509a511c577120aafa : SUCCESS in 50m 54s\n- python-tempestconf-tempest-devstack-admin-fips-centos9 https://zuul.opendev.org/t/openstack/build/5c6c4b3b5daf4d269c4195b0582135f3 : FAILURE in 20m 00s (non-voting)\n- python-tempestconf-tempest-devstack-admin-rocky https://zuul.opendev.org/t/openstack/build/37e2c3348e6840aa93036866017abc8a : FAILURE in 29m 15s (non-voting)\n- python-tempestconf-tempest-devstack-demo https://zuul.opendev.org/t/openstack/build/829d6fedfe3e45d9862e9d2bbe12dfb5 : SUCCESS in 32m 46s\n- python-tempestconf-tempest-devstack-admin-plugins https://zuul.opendev.org/t/openstack/build/75b737419c3e40b2b83005bbcfd431c1 : FAILURE in 39m 49s (non-voting)\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-admin https://zuul.opendev.org/t/openstack/build/4e8436d31324446892aefdc39837fa84 : SUCCESS in 40m 12s\n- python-tempestconf-tempest-devstack-enforce-scope-new-defaults-demo https://zuul.opendev.org/t/openstack/build/716919988e314e04bb47f086e1ef578c : SUCCESS in 41m 20s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/460880a4543e41e3992b5bb7297e47b4 : FAILURE in 5m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job python-tempestconf-tempest-devstack-admin-fips-centos9.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f6aeaca7fe7c9f3e5a3d1750b5564584b2d17de6","revisions":{"ecbdc9bdf7ebeace908cf03e046c75e30a181789":{"kind":"REWORK","_number":1,"created":"2026-07-12 12:55:05.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/40/996940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/40/996940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/1"}}},"commit":{"parents":[{"commit":"66ea243053f2b7a0abc015164baa4a338d4a2621","subject":"Detect cinder via block-storage service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/66ea243053f2b7a0abc015164baa4a338d4a2621"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-12 11:51:08.000000000","tz":0},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-12 12:54:45.000000000","tz":180},"subject":"Add Glance S3 store backend support","message":"Add Glance S3 store backend support\n\nAdd auto-detection of S3 stores via the Glance info_stores API and\na deployer-input template for S3 backend configuration. When an S3\nstore is detected, sets [image-feature-enabled] s3_store \u003d True in\nthe generated tempest.conf.\n\nChange-Id: I5194ea3d4a431710ef293250a2ae947965e2c509\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/ecbdc9bdf7ebeace908cf03e046c75e30a181789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/ecbdc9bdf7ebeace908cf03e046c75e30a181789"}]},"branch":"refs/heads/master"},"8893385bbb3e76c4930d177aa10482e1422aaac7":{"kind":"REWORK","_number":2,"created":"2026-07-12 15:11:05.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/40/996940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/40/996940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/2"}}},"commit":{"parents":[{"commit":"66ea243053f2b7a0abc015164baa4a338d4a2621","subject":"Detect cinder via block-storage service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/66ea243053f2b7a0abc015164baa4a338d4a2621"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-12 11:51:08.000000000","tz":0},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-12 15:05:25.000000000","tz":180},"subject":"Add Glance S3 store backend support","message":"Add Glance S3 store backend support\n\nAdd auto-detection of S3 stores via the Glance info_stores API and\na deployer-input template for S3 backend configuration. When an S3\nstore is detected, sets [image-feature-enabled] s3_store \u003d True in\nthe generated tempest.conf.\n\nChange-Id: I5194ea3d4a431710ef293250a2ae947965e2c509\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/8893385bbb3e76c4930d177aa10482e1422aaac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/8893385bbb3e76c4930d177aa10482e1422aaac7"}]},"branch":"refs/heads/master"},"f6aeaca7fe7c9f3e5a3d1750b5564584b2d17de6":{"kind":"REWORK","_number":3,"created":"2026-10-05 05:39:40.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/40/996940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/python-tempestconf","ref":"refs/changes/40/996940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/python-tempestconf refs/changes/40/996940/3"}}},"commit":{"parents":[{"commit":"66ea243053f2b7a0abc015164baa4a338d4a2621","subject":"Detect cinder via block-storage service type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/66ea243053f2b7a0abc015164baa4a338d4a2621"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-12 11:51:08.000000000","tz":0},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-10-05 05:31:41.000000000","tz":180},"subject":"Add Glance S3 store backend support","message":"Add Glance S3 store backend support\n\nAdd auto-detection of S3 stores via the Glance info_stores API and\na deployer-input template for S3 backend configuration. When an S3\nstore is detected, sets [image-feature-enabled] s3_store \u003d True in\nthe generated tempest.conf.\n\nChange-Id: I5194ea3d4a431710ef293250a2ae947965e2c509\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/f6aeaca7fe7c9f3e5a3d1750b5564584b2d17de6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/python-tempestconf/commit/f6aeaca7fe7c9f3e5a3d1750b5564584b2d17de6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
