)]}'
{"id":"openinfra%2Frefstack-client~614998","triplet_id":"openinfra%2Frefstack-client~master~I35e4facae2d35838a73cdb6100968e98b550ea43","project":"openinfra/refstack-client","branch":"master","topic":"min-tox-version","hashtags":[],"change_id":"I35e4facae2d35838a73cdb6100968e98b550ea43","subject":"Update min tox version to 2.0","status":"MERGED","created":"2018-11-02 05:54:47.000000000","updated":"2020-11-20 09:32:25.000000000","submitted":"2020-11-20 09:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614998-1605864745374-d1a6a7d1","meta_rev_id":"6d26048bfd7b2d2adb1dc4df50b977a919a484eb","_number":614998,"virtual_id_number":614998,"owner":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-20 09:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":1,"date":"2020-11-19 19:09:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-11-19 20:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-11-20 08:20:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-31 00:12:57.000000000","updated_by":{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},"reviewer":{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},"state":"REVIEWER"},{"updated":"2019-03-18 06:03:42.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2020-11-19 19:09:29.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-11-19 20:37:53.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-11-20 08:20:34.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-11-20 09:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a0c05bcc704f0b2c6d95999992e6edb16e6605e","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-02 05:54:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2ca84369c1e4cb4c26cc0bc2d0df42f8c89f899","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-02 06:46:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/614998/1/check/openstack-tox-pep8/d9211f0/ : FAILURE in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/98/614998/1/check/openstack-tox-py27/f8a5fa1/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/98/614998/1/check/openstack-tox-py35/6889d0e/ : SUCCESS in 3m 37s\n- refstack-client-devstack-tempestconf http://logs.openstack.org/98/614998/1/check/refstack-client-devstack-tempestconf/7003b39/ : SUCCESS in 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48dda8c5f06638439a76f700e8b48d606586a218","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-16 07:04:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88d252138f928032f664235f351fe9d033367899","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-16 07:51:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/614998/1/check/openstack-tox-pep8/ab70928/ : FAILURE in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/98/614998/1/check/openstack-tox-py27/7293e0b/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/98/614998/1/check/openstack-tox-py35/12113dc/ : SUCCESS in 4m 58s\n- refstack-client-devstack-tempestconf http://logs.openstack.org/98/614998/1/check/refstack-client-devstack-tempestconf/36d0b3b/ : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25d90fbe124b510319004b4be431d19190531346","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2019-01-25 07:26:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aab78161affda3c7e8c398436d908f4b8fa0de20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 08:15:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/98/614998/1/check/openstack-tox-pep8/cc153a2/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/98/614998/1/check/openstack-tox-py27/e843d51/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/98/614998/1/check/openstack-tox-py35/8644e49/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/98/614998/1/check/openstack-tox-py36/c59f40b/ : SUCCESS in 4m 53s\n- refstack-client-devstack-tempestconf http://logs.openstack.org/98/614998/1/check/refstack-client-devstack-tempestconf/afe2e36/ : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9e39636ecfabd2818106e6896d10d764a6ef9598","author":{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},"date":"2019-01-31 00:12:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1d4f1929bf4963a04e99b1a4a3ead1e7845a4da","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-03-18 06:03:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"713a171477b967f93cba7b1c461251766e229f23","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 15:50:58.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck ... yup, 2.0 is the bare minimum","accounts_in_message":[],"_revision_number":1},{"id":"8f6808318ceedc1a014bd877f329e67379559889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:51:10.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"ad938d97674f29b0f5bcd1ac0fa1512d144f3010","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 15:52:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"672f980919e809ab0b5578d27d4f0045705de807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:15:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a7f60aba574e9db8d7c72176f90644 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75f9b397b7b44fc483eee53e93b5e1d2 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e309b50d3a40178e4055fb1152f5ec : SUCCESS in 4m 06s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/97c56a600d254bdd920951b233968fc1 : SUCCESS in 1h 13m 04s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/41e9028be86f41c4a561e23fc657c225 : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":2},{"id":"92fb07c48c8139978d239adbbdf66754e0ee68fd","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 19:09:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9dc76cf6d089a7f0eb160211d60ee2febf6f6e14","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-19 20:37:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0da7d7def413a97509577ad0ade36e635a6e1ac6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-11-20 08:20:34.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e94b13ab8188f0a739518de197496ef07ca847f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 08:20:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ab7dcb18277f86c15a95ab8b819e8c5ba1b23c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:32:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cee382a8c1f54e108160e741110f62c0 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab591f4ddd1b49e49861244ac839562e : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/149b27a15ef7491fbed999834f622892 : SUCCESS in 3m 16s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/1f781b569d444559a7da6699c20e1965 : SUCCESS in 1h 05m 09s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/3f98de8a49e84773b0a6d91836955c69 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f18135586e0b07b2c22c6c069759a990552cc430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:32:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"474d3bbc4d57b39263972f9540f6529c0e6e8005","revisions":{"a6c24d91b0d04989344fbbeea566a7f360906fa8":{"kind":"REWORK","_number":1,"created":"2018-11-02 05:54:47.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/98/614998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/98/614998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/1"}}},"commit":{"parents":[{"commit":"ab8ec0a485f0e08b482c4338e717aa5ddc071ae9","subject":"Fix overrides and image_path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ab8ec0a485f0e08b482c4338e717aa5ddc071ae9"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-02 05:53:14.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-02 05:54:43.000000000","tz":0},"subject":"Update min tox version to 2.0","message":"Update min tox version to 2.0\n\nThe commands used by constraints need at least tox 2.0.\nUpdate to reflect reality, which should help with local running of\nconstraints targets.\n\nChange-Id: I35e4facae2d35838a73cdb6100968e98b550ea43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/a6c24d91b0d04989344fbbeea566a7f360906fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/a6c24d91b0d04989344fbbeea566a7f360906fa8"}]},"branch":"refs/heads/master"},"474d3bbc4d57b39263972f9540f6529c0e6e8005":{"kind":"REWORK","_number":2,"created":"2020-11-19 15:52:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/98/614998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/98/614998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/98/614998/2"}}},"commit":{"parents":[{"commit":"f069afca67d1e0b6a92557ea6856b804399bde82","subject":"Fix pep8, flake8 and unit test failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f069afca67d1e0b6a92557ea6856b804399bde82"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-02 05:53:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-11-19 15:52:22.000000000","tz":0},"subject":"Update min tox version to 2.0","message":"Update min tox version to 2.0\n\nThe commands used by constraints need at least tox 2.0.\nUpdate to reflect reality, which should help with local running of\nconstraints targets.\n\nChange-Id: I35e4facae2d35838a73cdb6100968e98b550ea43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/474d3bbc4d57b39263972f9540f6529c0e6e8005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/474d3bbc4d57b39263972f9540f6529c0e6e8005"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
