)]}'
{"id":"openinfra%2Frefstack-client~674679","triplet_id":"openinfra%2Frefstack-client~master~Ic569fa228d2481e0233130f392de7f74231d8e34","project":"openinfra/refstack-client","branch":"master","topic":"python3","hashtags":[],"change_id":"Ic569fa228d2481e0233130f392de7f74231d8e34","subject":"Bump guideline version to 2019.06","status":"MERGED","created":"2019-08-05 19:21:01.000000000","updated":"2019-09-17 13:02:58.000000000","submitted":"2019-09-17 13:02:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674679-1568725378041-332a2ac9","meta_rev_id":"edf0712e18b12673d8b856f571a36e0950db9936","_number":674679,"virtual_id_number":674679,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"date":"2019-09-17 10:34:15.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-09-17 13:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-17 09:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-16 10:26:06.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":1,"date":"2019-09-16 10:26:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-17 09:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 11:00:59.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-08-06 11:01:13.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-08-13 00:23:48.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-08-13 08:42:05.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-17 09:17:58.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-17 10:34:15.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-09-17 13:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e928f119967eaccd55e205bb2725476f34b7a0ed","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-08-05 19:21:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92706d35f0673b7c4334fde35ecac97e3b4a44c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/79/674679/1/check/openstack-tox-pep8/ee28c86/ : SUCCESS in 4m 25s\n- openstack-tox-py27 https://logs.opendev.org/79/674679/1/check/openstack-tox-py27/2324d41/ : SUCCESS in 5m 09s\n- openstack-tox-py35 https://logs.opendev.org/79/674679/1/check/openstack-tox-py35/083d1e9/ : SUCCESS in 3m 03s\n- openstack-tox-py36 https://logs.opendev.org/79/674679/1/check/openstack-tox-py36/e4907f4/ : SUCCESS in 4m 34s\n- openstack-tox-py37 https://logs.opendev.org/79/674679/1/check/openstack-tox-py37/abb1f96/ : SUCCESS in 5m 31s\n- refstack-client-devstack-tempestconf https://logs.opendev.org/79/674679/1/check/refstack-client-devstack-tempestconf/5d97c5a/ : SUCCESS in 1h 01m 37s\n- refstack-client-devstack-tempestconf-py3 https://logs.opendev.org/79/674679/1/check/refstack-client-devstack-tempestconf-py3/0bd6a2b/ : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":1},{"id":"467420f855b83b06c69c37156ac840ff72a9b06c","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-13 00:23:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cdc961cbd0f504213cb06025feb4b18150c5928b","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-08-13 08:42:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a88964e83f1a67f1c13fd6380083f918050d7ce","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-09-16 10:26:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4b40211ef663203eb9f2dd41aa5603d5bce5cd5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 11:34:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53d54a93d22435b85a490c1fb85102e : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d80563f483a4462e994be5a1acc72c2c : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/659b9261215a4cb5b3b10a18241da269 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25300a3658f246fcae491ee97abb32f0 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e89bdb562f0e4d9db683bcf4818ed04e : SUCCESS in 3m 14s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/1dd3ab36de1e43cdbc92a7998d077e5c : SUCCESS in 57m 12s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/4196863234d74811b67102797c4c297d : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":2},{"id":"48b9da8203baa2df3d0d81128d888b0316c8eea4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-17 09:17:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b92aed9ca11ff9086b28ff948b6a25da9b20a8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:18:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ba6c4cdcf708716f61171fd5879aeda4ce78182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:18:18.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"6ea5afc64bb372e90fa11a888a45c299acdd47f7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-17 10:34:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d6f88efa993a95ed783a9aab31c4bab7b67a6121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 11:59:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8991e216b508417cbc4a639249faeb55 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f70807433aec47e8a67d6db139d65969 : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b170bbf1a0134df39efff2693ea7f517 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81f7bbc9603143d4890efc93bf5ac2ad : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/431fa4f540a948daad1acc381826f1d8 : SUCCESS in 3m 33s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/50095d3bc1034f74bc46bbc7cdb6435c : SUCCESS in 1h 02m 54s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/3bb0e29304934cb0ab7d317bab8f9c98 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6a60a5f68412ac443accf2f77ea028faf5bdcd19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:00:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d12a08cfb4b01e4d5bd361b5eb131c8e610f88e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:02:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09438050950244b892c3c8ea30b97888 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a133aea205b4165b2d057d1676db36a : SUCCESS in 2m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29666e3b100c4c3cbf3856bb358dd078 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/247299d7c0fa411c802b2b17cb413640 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/437597f908e9416fb36a9d75f2b65ff7 : SUCCESS in 3m 56s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/1d17dfd1d2514a9fa4c439c8244d04be : SUCCESS in 49m 06s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/4b04ffc14f7c4cf68fc31a301d41cadd : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c43bd29fe2e0646536f9b6befffb0fbe749e0cf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:02:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6228a6a2979254a12b00fae45c4a4c74eb7810da","revisions":{"ec674161365ec44ba46d417dec269bcfce812564":{"kind":"REWORK","_number":1,"created":"2019-08-05 19:21:01.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/79/674679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/79/674679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/1"}}},"commit":{"parents":[{"commit":"7fa5cac36b6ab1a8813d61f584706f102235c197","subject":"Check python version when installing deps via yum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/7fa5cac36b6ab1a8813d61f584706f102235c197"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-05 19:19:55.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-05 19:19:55.000000000","tz":0},"subject":"Bump guideline version to 2019.06","message":"Bump guideline version to 2019.06\n\nJobs testing refstack-client use an old guideline version.\nThe patch bumps the version of the guideline to the newest one.\n\nChange-Id: Ic569fa228d2481e0233130f392de7f74231d8e34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ec674161365ec44ba46d417dec269bcfce812564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ec674161365ec44ba46d417dec269bcfce812564"}]},"branch":"refs/heads/master"},"6228a6a2979254a12b00fae45c4a4c74eb7810da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-16 10:26:06.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/79/674679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/79/674679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/79/674679/2"}}},"commit":{"parents":[{"commit":"bed0a57de94cdc53d6f295938e3304ce18081321","subject":"Check python version when installing deps via yum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/bed0a57de94cdc53d6f295938e3304ce18081321"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-08-05 19:19:55.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2019-09-16 10:25:10.000000000","tz":0},"subject":"Bump guideline version to 2019.06","message":"Bump guideline version to 2019.06\n\nJobs testing refstack-client use an old guideline version.\nThe patch bumps the version of the guideline to the newest one.\n\nChange-Id: Ic569fa228d2481e0233130f392de7f74231d8e34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/6228a6a2979254a12b00fae45c4a4c74eb7810da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/6228a6a2979254a12b00fae45c4a4c74eb7810da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
