)]}'
{"id":"openinfra%2Frefstack-client~708857","triplet_id":"openinfra%2Frefstack-client~master~I1c5abc8d0ee0b14d3b7294b0e77274c270c5055c","project":"openinfra/refstack-client","branch":"master","hashtags":[],"change_id":"I1c5abc8d0ee0b14d3b7294b0e77274c270c5055c","subject":"Initialize verbose flag properly","status":"MERGED","created":"2020-02-20 11:56:47.000000000","updated":"2020-12-11 12:26:43.000000000","submitted":"2020-12-11 12:26:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708857","meta_rev_id":"9e4d2d25783eb678f3780e551a6f08ba9e9757a6","_number":708857,"virtual_id_number":708857,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-11 12:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2020-11-19 15:39:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-12-11 11:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-11-19 16:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-12-11 11:21:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 13:00:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 15:39:05.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-11-19 15:39:22.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-11-19 16:08:57.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"}],"messages":[{"id":"b702b29bf1c3bddd937db603030cdbaee3055610","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-02-20 11:56:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93d23c7be48c5606cd9023a07c01bf56772f34a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 13:00:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef83cd77b8a7436f8c7a4a5a7dff641a : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d8cd245a403494ea0cb8aa09731dd7e : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5fb7fd4aa3334dc8a0ad726d93ba9185 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ed98409f6e4f69ad0ea8dd7b9afeca : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e672a66a2534a90a778163c4a7726f0 : SUCCESS in 4m 39s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/2b064387cfa74835b692f311962d6fa4 : SUCCESS in 55m 04s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/26bdb4e9fd014b54ab781165a7c418df : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bfc766b478d8df1170de6d3a80b81bf1db718310","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 15:39:05.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks, exactly, it can fail on this line (because we end up comparing None and int):\nhttps://opendev.org/osf/refstack-client/src/commit/f069afca67d1e0b6a92557ea6856b804399bde82/refstack_client/refstack_client.py#L562","accounts_in_message":[],"_revision_number":1},{"id":"b9ef23b77a9972f0168d5f1dd2239d5e0ca38bc1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-19 16:08:57.000000000","message":"Patch Set 1: Code-Review+2\n\nUh!","accounts_in_message":[],"_revision_number":1},{"id":"08c7745ae5afd75380233de81e92a08e3302ff1e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-12-11 11:21:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45a31a605c98bf241cffe0c9c1a0c6e889ca5788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:22:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b72d193d305d077a29216ee26cc97844c3013a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:26:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ec0368ae8294228aa904ae4df22a342 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/091ba1dc46cc4cb1bb96b625dbf2eb7e : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb9dd5e94234a4eb2766adc09a06c1c : SUCCESS in 3m 10s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/ad1c1df193ef413f9c569c5f08ef3476 : SUCCESS in 58m 34s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/7f26c12d616a4aa0a410cef732455ada : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9e4d2d25783eb678f3780e551a6f08ba9e9757a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:26:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d037611e99881fc27069de0a1943b7e1378a3450","revisions":{"d037611e99881fc27069de0a1943b7e1378a3450":{"kind":"REWORK","_number":1,"created":"2020-02-20 11:56:47.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/57/708857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/57/708857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/57/708857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/57/708857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/57/708857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/57/708857/1"}}},"commit":{"parents":[{"commit":"86c1618fee4bda2684d385b07151690c9ef0437e","subject":"Merge \"Bump guideline version to 2019.11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/86c1618fee4bda2684d385b07151690c9ef0437e"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-02-20 11:55:14.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-02-20 11:55:14.000000000","tz":60},"subject":"Initialize verbose flag properly","message":"Initialize verbose flag properly\n\nWhen running refstack-client without verbose mode enabled exception\noccurs, since the argparse argument is not properly initialized. Do this\nby setting default to 0.\n\nChange-Id: I1c5abc8d0ee0b14d3b7294b0e77274c270c5055c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d037611e99881fc27069de0a1943b7e1378a3450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d037611e99881fc27069de0a1943b7e1378a3450"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
