)]}'
{"id":"openinfra%2Frefstack-client~708859","triplet_id":"openinfra%2Frefstack-client~master~I5041d32be58a591db7eda7747ba2173527dffbf2","project":"openinfra/refstack-client","branch":"master","hashtags":[],"change_id":"I5041d32be58a591db7eda7747ba2173527dffbf2","subject":"Support DNF based OS","status":"MERGED","created":"2020-02-20 12:19:23.000000000","updated":"2020-11-20 05:11:23.000000000","submitted":"2020-11-20 05:11:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708859-1605849083056-795c03e7","meta_rev_id":"c70f0caae2d161b1136a3a91a8da268e3784434d","_number":708859,"virtual_id_number":708859,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-20 05:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2020-11-19 17:08:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-11-20 04:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-11-19 20:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-11-20 04:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 17:08:26.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-11-19 20:57:28.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-11-20 04:19:35.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-11-20 05:11:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae5ff29a02bbf2014904abed1c0d9d59c7bffaf9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-02-20 12:19:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6a9a4e1fc8f5a5ac1f2c807ad5f4c9641b07ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 13:22:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34303b00cf04d2da81a92840a503731 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f64becac052c4aaaa515ca5f98c4dc3a : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0aef74889c994a0397c90ce8a9e7b896 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45684c0c884640ff94a5f93d22948e42 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e21e4aa4dc3f4af498c29bdc9f76c28e : SUCCESS in 6m 41s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/98fa40e5530c4a129fd960514a27ba12 : SUCCESS in 56m 59s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/05eb0af967734d4bbf1d832a5dfd9b67 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b64336b61e08def97a6b4d71c1e8f6a2110a449a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 17:08:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"121039b9492264871b64837d4ee4507b0ca7e51d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 17:08:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"796788d38fef149459c038763548d86291eb26b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98ee86f089e24ce79ccc5b89fbccf0a0 : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ecb30bd12f04f30915a3bac8aa63f1a : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ec7430e939432b99f2681faa5d2ea6 : SUCCESS in 2m 59s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/ac6990f8ec6a42048ef9f8393ae3c7fb : SUCCESS in 1h 19m 32s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/ac6038a6039e4a9690bc50cc75c1ee9c : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d2eb800f02a16d5d48f5c1d157401f95cd0dfc85","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-19 20:57:28.000000000","message":"Patch Set 1: Code-Review+2\n\nIt seems it should work properly on CentOS 8/RHEL 8 (which ships dnf but the documentation still advise people to use the yum alias).","accounts_in_message":[],"_revision_number":1},{"id":"9b06ad7cb21a824a356fdb2bfedfd36301b56255","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-11-20 04:19:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a0915261d272a185847bdaa689b1aad2ad48473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 04:19:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31bcab3393a8cc600f550d5443196ffc97a807c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 05:11:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf8a6db577446ef968d5bbd4a54ccbe : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/009a917ff77d4640913c6163d8c45f6e : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366c5927dd9d4a5abf073c5f23a12d2d : SUCCESS in 3m 45s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/7313d044d66d4ef28b8dbed2f0de5fdd : SUCCESS in 46m 51s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/fba8a9dff0d3481c9231bf7fcb2d5e71 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c70f0caae2d161b1136a3a91a8da268e3784434d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 05:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4e4da0ffc8e506520410e419c3c53ee3d0d7b8e","revisions":{"d4e4da0ffc8e506520410e419c3c53ee3d0d7b8e":{"kind":"REWORK","_number":1,"created":"2020-02-20 12:19:23.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/59/708859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/59/708859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/59/708859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/59/708859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/59/708859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/59/708859/1"}}},"commit":{"parents":[{"commit":"86c1618fee4bda2684d385b07151690c9ef0437e","subject":"Merge \"Bump guideline version to 2019.11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/86c1618fee4bda2684d385b07151690c9ef0437e"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-02-20 12:18:06.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-02-20 12:18:06.000000000","tz":60},"subject":"Support DNF based OS","message":"Support DNF based OS\n\nSince some time RedHat based OS are moving from yum to dnf. Extend check\nof the supported underlying package manager to add \"dnf\" command\n(interface is same as yum).\n\nChange-Id: I5041d32be58a591db7eda7747ba2173527dffbf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d4e4da0ffc8e506520410e419c3c53ee3d0d7b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d4e4da0ffc8e506520410e419c3c53ee3d0d7b8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
