)]}'
{"id":"openinfra%2Frefstack-client~738584","triplet_id":"openinfra%2Frefstack-client~master~I7cce77dae30b1331c093a560946dc5ff31eb986a","project":"openinfra/refstack-client","branch":"master","topic":"use-opendev","hashtags":[],"change_id":"I7cce77dae30b1331c093a560946dc5ff31eb986a","subject":"Replace GitHub repositories with OpenDev repositories","status":"MERGED","created":"2020-06-30 11:18:03.000000000","updated":"2020-11-19 16:48:12.000000000","submitted":"2020-11-19 16:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738584-1605804492941-6db76d6a","meta_rev_id":"1565e2afef1a7df2303ddbafea9028420f7445b4","_number":738584,"virtual_id_number":738584,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"date":"2020-11-06 10:57:38.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 16:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2020-11-03 15:22:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-11-04 12:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-07-02 07:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"date":"2020-11-03 15:24:03.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-11-04 12:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 07:30:08.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-11-03 15:22:07.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-11-03 15:24:03.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-11-04 12:31:02.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-11-19 16:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77a615f9d842fb6c5f607de379f963463b6221ea","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-30 11:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12922de12b7a5b8e80154e3cc595b5d4f60c7c3f","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-30 11:56:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a61c6276a2117168b9f2b172fd98793f8f99ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 12:34:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/619d911f49ca4842b304405b2438f8fc : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33e5b4c02c594aaf9a394eb437525e0c : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71f6c2d8b924907bce5c0834030a181 : SUCCESS in 3m 52s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/b6c8035c50fc414692bfa5e849e37d9a : SUCCESS in 34m 48s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/4c97d1e34328457eafa57a2ee4232859 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":2},{"id":"268fb42b8dd5dfbd0bec29782b75580370d2c56b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-02 07:30:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1886acfeaec0b8ba9b5a906e60499114882a89f4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-03 15:22:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b1b5dbbb00695631319d09f11f6c9b1dc64ef719","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-11-04 12:31:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f1257282737a7a3b5e37b4a806144a1e99df1fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 12:31:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76631db90a2e3d2ed276b64ec6c5f4728b1796dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:37:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b339f083d534f60a35ac50fbcfc4582 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50bce7cb3953454ea242679529cba2af : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eed22dcf67c4abe8229ec85029056a9 : FAILURE in 3m 05s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/ea55d6b0341e429d896ae41ab61238e1 : SUCCESS in 58m 22s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/1a8a133d6ac54f608c38ea1cc2386817 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f809241faef74fe9ce8189685c529896a49621ad","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-04 13:40:54.000000000","message":"Patch Set 2:\n\nwe need to adapt to the latest flake8 changes. Any takers?","accounts_in_message":[],"_revision_number":2},{"id":"6a23ecf841ea6ae8935e6e48a11c7c638b95c6ae","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-11-06 10:57:38.000000000","message":"Patch Set 2:\n\nWe will take care of it and open a review at the beginning of the next week.","accounts_in_message":[],"_revision_number":2},{"id":"b1aef325e78a1fc33039bce285f7a8beceddabde","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-18 16:55:15.000000000","message":"Patch Set 2:\n\nthis will fix the pep8 job -\u003e https://review.opendev.org/#/c/763188/","accounts_in_message":[],"_revision_number":2},{"id":"7ba1d602efe9d615ffcbbf0c87ab450fbf14610a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 14:38:49.000000000","message":"Patch Set 2:\n\nrecheck - https://review.opendev.org/#/c/763188/ got merged","accounts_in_message":[],"_revision_number":2},{"id":"58198b017a219767d5960b3e73b0065c2c4795db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f20886bfb14a52a75fe17a9b05041b : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac0484a36d924749aa48cf5108c5d403 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866cbdbdfd5341feb19edfd4da23db8a : SUCCESS in 3m 13s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/a26e6d3f33dd445781d6e90d6fc307e3 : SUCCESS in 1h 01m 28s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/b53a8df35c0344b18c6a3bae041d917a : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5d57a45539badd9d6ca176e820f4e5ffcee7dc73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:46:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf42b6d033ee10ff628d6e633b80355d2d746a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 16:48:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f73485ea5fc548e4bf033634403bfbed : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12ffb6d1f62740fb953edb48f25c6a02 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6f162293b84b42b23edf53591c171e : SUCCESS in 3m 12s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/5d19acd83c504b1cb6f6f1b1855c183a : SUCCESS in 59m 17s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/9025340ef7204258a853644cb76d0b54 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"318d0f39e24a664917e505ddf647a680bf2dec02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 16:48:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a6ceb8acfb19dc58282206f855bdd727a262deb","revisions":{"85b4f5fd8c62852eecf7ef68d3c40fa592248ee7":{"kind":"REWORK","_number":1,"created":"2020-06-30 11:18:03.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/84/738584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/84/738584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/1"}}},"commit":{"parents":[{"commit":"a2b4ce30b8710e199f22e9de61b51c5642d5581e","subject":"Update jobs for interop repos renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/a2b4ce30b8710e199f22e9de61b51c5642d5581e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-30 11:17:12.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-30 11:18:01.000000000","tz":120},"subject":"Replace GitHub repositories with OpenDev repositories","message":"Replace GitHub repositories with OpenDev repositories\n\nChange-Id: I7cce77dae30b1331c093a560946dc5ff31eb986a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/85b4f5fd8c62852eecf7ef68d3c40fa592248ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/85b4f5fd8c62852eecf7ef68d3c40fa592248ee7"}]},"branch":"refs/heads/master"},"4a6ceb8acfb19dc58282206f855bdd727a262deb":{"kind":"REWORK","_number":2,"created":"2020-06-30 11:56:46.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/84/738584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/84/738584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/84/738584/2"}}},"commit":{"parents":[{"commit":"a2b4ce30b8710e199f22e9de61b51c5642d5581e","subject":"Update jobs for interop repos renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/a2b4ce30b8710e199f22e9de61b51c5642d5581e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-30 11:17:12.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-30 11:56:37.000000000","tz":120},"subject":"Replace GitHub repositories with OpenDev repositories","message":"Replace GitHub repositories with OpenDev repositories\n\nChange-Id: I7cce77dae30b1331c093a560946dc5ff31eb986a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/4a6ceb8acfb19dc58282206f855bdd727a262deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/4a6ceb8acfb19dc58282206f855bdd727a262deb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
