)]}'
{"id":"openinfra%2Frefstack-client~763188","triplet_id":"openinfra%2Frefstack-client~master~I51558ee9c808e03b4164d0832fc525c5556dddd1","project":"openinfra/refstack-client","branch":"master","hashtags":[],"change_id":"I51558ee9c808e03b4164d0832fc525c5556dddd1","subject":"Fix pep8, flake8 and unit test failures","status":"MERGED","created":"2020-11-18 14:16:47.000000000","updated":"2020-11-19 14:34:05.000000000","submitted":"2020-11-19 14:34:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763188-1605796445331-b16e9fe7","meta_rev_id":"4772a1af2f766113a8ae7a585e8c78037acb741d","_number":763188,"virtual_id_number":763188,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 14:34:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":2,"date":"2020-11-19 04:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-11-19 13:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2020-11-19 13:34:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 04:09:45.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-11-19 13:34:46.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-11-19 14:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03bdc2bc8c673f4d0a97574f94fa5759d2905849","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-18 14:16:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd902e1f78210998e02ba8114ce98d57b0a8f727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 15:23:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff57845a74b34eaabdc08b01fe4b367c : FAILURE in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb4d956bf2cb4879b29627c3bdc6cd90 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bcd964788974b78a81164ac5eb0e12f : SUCCESS in 3m 04s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/a408cbc7f57e43fa973c367e00511b02 : SUCCESS in 1h 00m 23s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/94c820e2559b43e99cc1884cba3d1517 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5b95423f528e03d4a726fdc6a6d4180192611c95","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-18 15:42:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8ababb554ebcc96781b165f2538f2dfbf12443c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:50:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b96b515d1045c6a65ef6c4685dbfca : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f0ebb24135e4087befa9a1bde419bc8 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cc5695703f4f248d9f45a3fcd7edd2 : SUCCESS in 3m 00s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/4d215d700b8549fabe5f84c5a9e56d53 : SUCCESS in 1h 02m 19s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/dd21b1c0e3b34df2bd791ec7947f79ac : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5e03a8d4baf666ad277b2ebdef2088670ee568db","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-11-19 04:09:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"529ec28cb9c9ef86c15437ffe3a676b9578e7c2f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-19 13:34:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a068be03c98713e93e5ac419f6de172b3064b136","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-19 13:34:52.000000000","message":"Patch Set 2:\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"61c9b9108a929aba18b3b1bfa88a852b76190473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:37:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ea5adbe2f6f2fd0edfe406278a0dbb6fd960a9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:34:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4772a1af2f766113a8ae7a585e8c78037acb741d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:34:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37d2c1a54c244221831e1cb1d31eab83 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8c5fbf96a3c4bf3a4ea3ed1743376fb : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bef63a7bd254d16ac6bef99f3b4073f : SUCCESS in 3m 13s\n- refstack-client-devstack-tempestconf https://zuul.opendev.org/t/openstack/build/60fae129daae42a09fd9f2984298f206 : SUCCESS in 53m 29s\n- refstack-client-devstack-tempestconf-py3 https://zuul.opendev.org/t/openstack/build/480cfad72e744279b2a630a5f8955971 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f069afca67d1e0b6a92557ea6856b804399bde82","revisions":{"5965b31ac1d6c40cc190261a45f9bb893c7d6fe3":{"kind":"REWORK","_number":1,"created":"2020-11-18 14:16:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/88/763188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/88/763188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/1"}}},"commit":{"parents":[{"commit":"a2b4ce30b8710e199f22e9de61b51c5642d5581e","subject":"Update jobs for interop repos renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/a2b4ce30b8710e199f22e9de61b51c5642d5581e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-11-18 14:13:26.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-11-18 14:15:23.000000000","tz":0},"subject":"Fix pep8, flake8 and unit test failures","message":"Fix pep8, flake8 and unit test failures\n\nThe patch edits the code so that it complies with W605 rule\nwhenever it makes sense. W504 rule is added to a ignore list.\nHigh version cap of pyflakes and flake8 packages is increased\nas well.\n\nChange-Id: I51558ee9c808e03b4164d0832fc525c5556dddd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/5965b31ac1d6c40cc190261a45f9bb893c7d6fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/5965b31ac1d6c40cc190261a45f9bb893c7d6fe3"}]},"branch":"refs/heads/master"},"f069afca67d1e0b6a92557ea6856b804399bde82":{"kind":"REWORK","_number":2,"created":"2020-11-18 15:42:38.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/88/763188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/88/763188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/88/763188/2"}}},"commit":{"parents":[{"commit":"a2b4ce30b8710e199f22e9de61b51c5642d5581e","subject":"Update jobs for interop repos renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/a2b4ce30b8710e199f22e9de61b51c5642d5581e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-11-18 14:13:26.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-11-18 15:42:33.000000000","tz":0},"subject":"Fix pep8, flake8 and unit test failures","message":"Fix pep8, flake8 and unit test failures\n\nThe patch edits the code so that it complies with W605 rule\nwherever it makes sense. W504 rule is added to a ignore list.\nHigh version cap of pyflakes and flake8 packages is increased\nas well.\n\nChange-Id: I51558ee9c808e03b4164d0832fc525c5556dddd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f069afca67d1e0b6a92557ea6856b804399bde82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f069afca67d1e0b6a92557ea6856b804399bde82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
